)]}'
{"id":"openstack%2Frequirements~831731","triplet_id":"openstack%2Frequirements~master~Iba36a74567394faacb956bf5dbeb33688c8f187b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iba36a74567394faacb956bf5dbeb33688c8f187b","subject":"update constraint for kuryr-lib to new release 2.5.0","status":"MERGED","created":"2022-03-03 11:15:25.000000000","updated":"2022-03-03 17:08:24.000000000","submitted":"2022-03-03 17:06:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831731-new-release","meta_rev_id":"c98ae8b6bae0279e9aac65aed99965ec38e4cef5","_number":831731,"virtual_id_number":831731,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 17:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-03-03 15:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-03-03 15:17:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-03 11:43:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-03 13:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-03 15:17:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c0d5411c22b4532a409d2ddbe5c95b0a0151791c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-03 11:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb0230066c35c189483f05891285e6f58b324d84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:43:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3b68c5f8dca54ab38a845299971f1068 : FAILURE in 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"dec62711d894afeb40a1b39f14f9b67c7ef139b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:02:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a662d4dec8a84db7a98c40efba500cac : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3970431415ab4c26a3934392bc01e7c1 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ad5ed26de0c454bb63a37d4a4f7e4a7 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe3db77d81d44ec7a34e92ebb881fa91 : SUCCESS in 7m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f23fd366159a4aed93c9419199eceb77 : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28709e840fa7402ea85d8cfbe24322c8 : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/473ded1009c844ebbec512e1df8369cf : SUCCESS in 13m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94edc05017244c589a50b733f6070bc6 : SUCCESS in 12m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/083bbacdf0494993801864cece79ed16 : SUCCESS in 8m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/06301f9a9acd44c9b64973cb2ec5ae65 : SUCCESS in 12m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/196ecbe63b044b1fab3522d458c93363 : SUCCESS in 7m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/25152bf1aba34c63bb65893c6291a722 : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/43fc0e5a626c4c3c93a63e5b794f5c1e : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/870aef3b13ba40caac80008485f6dca5 : SUCCESS in 11m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/316e0dfce7304147a0d3564d28beac87 : SUCCESS in 10m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b275050ee9d1490e8c33039d79e42f40 : SUCCESS in 13m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a228457c2acd4a038caa4a51bcb0372f : SUCCESS in 6m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ca503bdd55a14ef78397c8915fd8ef1d : SUCCESS in 10m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8fb879ef76b34c9dada9d08fbe131c51 : SUCCESS in 7m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/867a6cfd5ad0423da0c2d92d397da7b2 : SUCCESS in 24m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8035571823a94b85add1414fbf7f6b9f : SUCCESS in 12m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3e0f11da93a044b3af583b9243c89977 : SUCCESS in 7m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f76ea4d9247549b096a336abcb5b5796 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e4359af1c044e6193311fab37f7c7cc : SUCCESS in 29m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fdf49aa005204fd294df867c87132939 : SUCCESS in 8m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a7366788dfb045048c91223987c80717 : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc1b75ff1f6047508defbd3c3739aa0e : SUCCESS in 9m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6c3f1f62896a41f8a32c480f723d5ff9 : SUCCESS in 18m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2fd0ee4114c47e2874d85ae7edea53f : SUCCESS in 50m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d718013cd4f74e9c9d82e453758cb607 : SUCCESS in 1h 31m 49s","accounts_in_message":[],"_revision_number":1},{"id":"786c261ed0500a0e89805cf258b6f76e2b9c4808","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-03 15:17:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c51ad27ebe71b433fb2ad6bd36796169e672bf59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:18:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e52bfbbe767cc604837dd6707c1a466f9e7d57c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 17:06:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3519f22661594b00a16650394979f115 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/945ed46c802a4a0a88d7b255c759c079 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d6ef9d14454cb2ab7129b499a21ffe : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27ade15ff49b44c98abfe2c944907ab8 : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95690452c092437aad865d55de982d76 : SUCCESS in 6m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8202b8064ee42cda0dac53e564da9e3 : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f73b634c6ef44a7be7960a5005c02f7 : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f13dae6192d44b5fb65c37ec2ecdc6d5 : SUCCESS in 13m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/da31c3f9a1bf4239902c18718e315c45 : SUCCESS in 13m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f97670a95e6c4a75bace8d078f827820 : SUCCESS in 9m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c336742cad8049e88713e3c1f79973c5 : SUCCESS in 9m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d6b8aab8f069440f9280a04bdd9a658d : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71356b872f414866a2b615e7ba1478ba : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ced84a534f1a493ca82c50b0bc39cdc5 : SUCCESS in 10m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6683757569044445830b308c65f53950 : SUCCESS in 15m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9d6fb1883abd4dc3881f60680322dd83 : SUCCESS in 6m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5ce3b1e0d7554fbd88a5e5bc12f5207e : SUCCESS in 14m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4e928ec4eedc45d5b652162dab38be9a : SUCCESS in 7m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/067c56b587884dafab5cad8ab6710d74 : SUCCESS in 32m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/810e781273414947b920306d55a1c0e4 : SUCCESS in 15m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3cf40738b9c748ecb455d51a995c97c9 : SUCCESS in 7m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/040834787a6743f483627dac4dfa2537 : SUCCESS in 5m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bad8a00e3cfe4338b67f43e51356e13b : SUCCESS in 20m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a322aad7819144fe979a2fd02d554c00 : SUCCESS in 7m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6a6ff1bac97041ddbc4dd1105c4c34d8 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37859d6787f34e8ca6a816b7af210251 : SUCCESS in 9m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a2ddcef559b14b4094bc22e972152921 : SUCCESS in 14m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bcce725202b45088c0e0abf573c9080 : SUCCESS in 1h 04m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08bbffa7a43145d7a394084bfa3752d0 : SUCCESS in 1h 43m 04s","accounts_in_message":[],"_revision_number":1},{"id":"05073b5d99e20db669296e2617cefcc08639246f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 17:06:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c98ae8b6bae0279e9aac65aed99965ec38e4cef5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 17:08:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102a7271e423453db982e131055ff13c : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"804ffbf3aca9393fec7e44ad49753f47c6c3933c","revisions":{"804ffbf3aca9393fec7e44ad49753f47c6c3933c":{"kind":"REWORK","_number":1,"created":"2022-03-03 11:15:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/831731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/831731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/831731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/831731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/831731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/831731/1"}}},"commit":{"parents":[{"commit":"48fb5c24764d91833d8ca7084ee9f183785becd6","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48fb5c24764d91833d8ca7084ee9f183785becd6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-03 11:15:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-03 11:15:24.000000000","tz":0},"subject":"update constraint for kuryr-lib to new release 2.5.0","message":"update constraint for kuryr-lib to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I8b8f25c38835ef26771e6c64539f3e6940a571e4\nmeta: release:Code-Review+1: Maysa de Macedo Souza \u003cmaysa.macedo95@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iba36a74567394faacb956bf5dbeb33688c8f187b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/804ffbf3aca9393fec7e44ad49753f47c6c3933c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/804ffbf3aca9393fec7e44ad49753f47c6c3933c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
