)]}'
{"id":"openstack%2Frequirements~832230","triplet_id":"openstack%2Frequirements~master~I14669a6dd9cc6c4ae9a4806f282c1e5e1c582b6f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I14669a6dd9cc6c4ae9a4806f282c1e5e1c582b6f","subject":"update constraint for sushy to new release 4.1.1","status":"MERGED","created":"2022-03-07 13:54:25.000000000","updated":"2022-03-07 19:12:13.000000000","submitted":"2022-03-07 19:09:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832230-new-release","meta_rev_id":"83cb831ccff64d6daa13ce519d7c75eaf069f00f","_number":832230,"virtual_id_number":832230,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-07 19:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-03-07 17:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-03-07 14:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-03-07 17:16:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-07 14:11:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-03-07 14:15:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 15:22:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-07 17:16:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6f19323adec79ee139e00829f073fb4574d1cd1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-07 13:54:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"527d35c95b20e4056260332572cb2a5609e30433","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-07 14:11:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e05fa39d4f0a13b6ac765f5f0ac0f52ff8fae66","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-07 14:11:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cc6d98a6fdd3ff55f175dfbef7ff5c3bbe4f460","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 14:15:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d5ea98d275f24966bb3098e00d0287d0 : FAILURE in 17m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b8ff681768d97dbc3b1788c872e02515de7173f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 15:22:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091b00eb06d7439f8273c008415a7200 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1fff0f9383546de905f847e787f30bb : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70d5a593718b4677a7ba0bbd6538d9b2 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a514e59b7bc4185b6659f7df6e918ae : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffa0504d2594416d9e5c8af9f49b10d2 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32b0353ca0b54004b35ce84eaafc4c66 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/80277bcc36a84e00a7438ff48520ae51 : SUCCESS in 15m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3a8e5df2fa5479c9818d43e2a9d4c25 : SUCCESS in 10m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4349463ab6124b2ab3746a50a318a905 : SUCCESS in 6m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c332034197fd45fb859ea203e98f6641 : SUCCESS in 14m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0e8b900a07f147768bd3826c3dc2044f : SUCCESS in 6m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9fbeafaf1192473fa762079544e3bc35 : SUCCESS in 7m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6c0f03f6361a4c9184f6a7e6548037e8 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44ea5e1cef864849acd1531adcc0f36a : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0b1007c7f9d34c10bc06cb6ea8e33713 : SUCCESS in 10m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e759aa74562247f797e691a34c12c2d9 : SUCCESS in 13m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6cf6bcc1c21f4542a16c0257a4fd94d4 : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e38cf261444848569ef3153a8221b1d0 : SUCCESS in 14m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0c6e9fef7f854a37a25a68358e87bede : SUCCESS in 6m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bbe5475d22b847428946d44122d4f220 : SUCCESS in 31m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a4b73f17d74c4ef5a73ffd6fb1f92105 : SUCCESS in 13m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/776eace1c1ff45b38f32a82b08930ef4 : SUCCESS in 7m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/884daebe3a1d443b94e3f70a3dd89e20 : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ada8adf0a9545dd9668dd4d4e20129b : SUCCESS in 22m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/375da7bf91d04282988846fccbcd1117 : SUCCESS in 7m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cdfe62f2df1d4105962e9417cf2ae5a7 : SUCCESS in 7m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad25e72b6a724d00a3bcb9da7fe0a9f9 : SUCCESS in 8m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8b60cd82f648453daf3e3d1eb681bd7b : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91d516b849fd4189abe6c9c5103772be : SUCCESS in 55m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7ce85e592d24bbaa46ef2c81899c51c : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"92965464750784ef4635c7c82ca0d13829902b0d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-07 17:16:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"368ab08b804375cc601a23af354988e11ab2d7cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 17:17:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7262dd11c58f3e64271644eee7eb4caa1ae2615b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 19:09:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888c8764b3cb4711bce17694400bbb48 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1085114d6340c0a2584b4860691333 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6da83193beb448380d1f566780f7c30 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1610888bcfcd433587c0d59ebfc11664 : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8894ca4af2446779a783492b37cad0c : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51ee5f42052d4be381e470058d4b6811 : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/47c2f4902d9842c9962ce4ed4311d73d : SUCCESS in 10m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/708a00093bca483cad3b57237da95bc3 : SUCCESS in 11m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e84f59ddb28941719158b89925351978 : SUCCESS in 14m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6c0d5e8e7b934c05a141603c0b11820d : SUCCESS in 6m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/708ade1bef1b4c1f9b8cc27700f47a19 : SUCCESS in 7m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4ed4a585afc54dbca2ed016f7db7df4b : SUCCESS in 14m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cbf497e50cdc484fb2b2941feaedd3d1 : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49ba3d8138d843029447e0da9a09c0d5 : SUCCESS in 10m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ba3ab907f77e4165a7a5143d98f8a16a : SUCCESS in 13m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c009b57f6bde44038ca12c383a7d8f19 : SUCCESS in 5m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2ec5905f248a41648d4c97533b21fb09 : SUCCESS in 11m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/24c6452700e04a63889be7f05fcd4ae4 : SUCCESS in 6m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/260953ed2bd749f29b0d53ca05d998c1 : SUCCESS in 25m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eb09fc576e6d47858f0d824984756c7e : SUCCESS in 14m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0903c5dad4b144da9e15e1ffb4ba9b2d : SUCCESS in 8m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca392d713f954c05bbb31886b61e2f87 : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74ad87a91a574aefb55e407b0cddf2e2 : SUCCESS in 21m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/724e13e54c8b40b0b90fc82f7e76b0bf : SUCCESS in 7m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ddaea4a9548e40ff92e3eebe122ff190 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/870930ae98b941eb8a8d20a4667ca252 : SUCCESS in 9m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8ab934e8e4944cedadae6d194190fc82 : SUCCESS in 15m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e294667817444bf930bcb4e2356571a : SUCCESS in 48m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e52be899df184f4ab6dad4033abcda37 : SUCCESS in 1h 31m 55s","accounts_in_message":[],"_revision_number":1},{"id":"73d396a03f163bea46b8151d24237ba6c8c65e5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 19:09:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"83cb831ccff64d6daa13ce519d7c75eaf069f00f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 19:12:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a00d7eb386b447b9d00febcbb8ccd4c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62493f028449034a11c34cbc70e5e55f3bd44275","revisions":{"62493f028449034a11c34cbc70e5e55f3bd44275":{"kind":"REWORK","_number":1,"created":"2022-03-07 13:54:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/832230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/832230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/832230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/832230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/832230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/832230/1"}}},"commit":{"parents":[{"commit":"be1172b162a357cace0e115ffdcf5fa4634a6215","subject":"Merge \"update constraint for oslo.context to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be1172b162a357cace0e115ffdcf5fa4634a6215"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-07 13:54:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-07 13:54:24.000000000","tz":0},"subject":"update constraint for sushy to new release 4.1.1","message":"update constraint for sushy to new release 4.1.1\n\nmeta: version: 4.1.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Commit: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Change-Id: I94421fe0cf086c6d84f366be99e6b834a1a86f04\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+1: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I14669a6dd9cc6c4ae9a4806f282c1e5e1c582b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62493f028449034a11c34cbc70e5e55f3bd44275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62493f028449034a11c34cbc70e5e55f3bd44275"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
