)]}'
{"id":"openstack%2Frequirements~832277","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Id065184c67b5393cc3cadca0267f5449dc6e0b89","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Id065184c67b5393cc3cadca0267f5449dc6e0b89","subject":"update constraint for os-brick to new release 4.0.5","status":"MERGED","created":"2022-03-07 16:03:02.000000000","updated":"2022-03-07 19:09:56.000000000","submitted":"2022-03-07 19:09:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832277-new-release","meta_rev_id":"892ceb8258c834c1eaabe656f3d1057b4f1fa425","_number":832277,"virtual_id_number":832277,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-07 19:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-03-07 17:43:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-03-07 16:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-03-07 17:43:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-07 16:04:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-03-07 16:39:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 17:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-07 17:43:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"67ea8cd4cba8b7579545b75001ec09464a44a9d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-07 16:03:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba129f4926f9fb97de2dd994527c6840ef4b00e9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-07 16:04:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f0a87ef66f7a45a3b3f9180fe38df41851cd6f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 16:39:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c28e5e390728410c81e5d2b46414f8dd : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"be594ba3481cb9ca2364799dbfdb0e90d33d56db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 17:37:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3027aad75bc644a0a079d9ba5958158e : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/134f886c92604dadb264f0e08e72dd00 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9d123b31fd146bb94474e6db1c24da2 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3d7058a62294dceadf102dcd2c79705 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ac3fd94d106482c946b91a7b7231dfd : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/21fdf7deb2624e78974e835ab424053b : SUCCESS in 14m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/118bbc4dc9f348129d01fe02f37e76b8 : SUCCESS in 15m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ecdcd73f2398451b8c3bfb1d22127e14 : SUCCESS in 12m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa5f478a0b4f4ceaa046a04e72f3463e : SUCCESS in 8m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba00bc72b8074724bbec5024a2a2cf22 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb1f47525a474309bbdc2f15457b2877 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/30c4ea9e443e485b88ae8f348e811353 : SUCCESS in 9m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7af656d009084f6aaa017fcfc5ff8116 : SUCCESS in 14m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/751c7bb25fa74ba6b508085b45158117 : SUCCESS in 27m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/546e5566d1f14d398888c2dc47d39f17 : SUCCESS in 14m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bdfbb45ba3c343bcaeac1f0dac9715a6 : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97e3174e7f0a4915b2b7bc8fb9eafb76 : SUCCESS in 17m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5f8a1db9d4c848a4bf2dd8721fe12f40 : SUCCESS in 8m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86f09b5db43b493bbcfc97156721e466 : SUCCESS in 8m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a2915e44f0be45b4bdeb851cf638c8d7 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0ae5a1d073840758f13b3bb064b680f : SUCCESS in 54m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68597d63e93744e6b168635c922c0d4e : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":1},{"id":"72301e2e032fcb93c3be0b1bc4a4aef3f4b8793c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-07 17:43:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b6a73e2604da25ed11f1badfc0d7b526f3cec5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 17:44:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24ee38d00cfe9c53c2e8a733370fdaf7b49883c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 19:09:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75e7534a90fb4255be40af777b6e7a24 : SUCCESS in 20m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/159c04a7542441638cf50daf35fae954 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ccdcf334d504d48b9f05f52be7f1738 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2775aa3e1694538a1c93f33742f51c9 : SUCCESS in 5m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94666a02c7894871a75544d6a083a910 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01fa17b3626f47fb877e13d0bf832ac6 : SUCCESS in 11m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/924e32304e684af18186de65b085b2b3 : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2715867697b47f6acdeb5603f62bf9a : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b9f1b165bcb47fc9ed52caa3d9ba5b8 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a7247f3728304ea6a93f27631cff47c0 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e0f558a2e4042b0b9ebc4216297e2fe : SUCCESS in 10m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0e301f3f6ca04151adc6df7a4f53a476 : SUCCESS in 8m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c016292445f4760aebc95b4cbe17e54 : SUCCESS in 14m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc00802ad9864fa1a506e98e34ba1188 : SUCCESS in 26m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8a52203acd9743c79f9f42dc1875f177 : SUCCESS in 13m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1d0abb16da740bc99199b040e54d193 : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/713622f1706243daa1f1bf596927e262 : SUCCESS in 19m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c31189a952a4ccdbafdd75bd1edf3a7 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5bc94450d11242bf91f698dbb6a4d04b : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cb77e20544924c6e804ead25a673d43a : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ca205cbf5ac4d75816e7207faaa191c : SUCCESS in 44m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7661718f351246a8aa0b3a5f03039c72 : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":1},{"id":"892ceb8258c834c1eaabe656f3d1057b4f1fa425","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 19:09:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"465d12d1ac30987a981609407d84a94caf3b519a","revisions":{"465d12d1ac30987a981609407d84a94caf3b519a":{"kind":"REWORK","_number":1,"created":"2022-03-07 16:03:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/832277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/832277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/832277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/832277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/832277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/832277/1"}}},"commit":{"parents":[{"commit":"8b422b75cbf4b2e2441a1baedd426c60116a964f","subject":"update constraint for oslo.utils to new release 4.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b422b75cbf4b2e2441a1baedd426c60116a964f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-07 16:03:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-07 16:03:00.000000000","tz":0},"subject":"update constraint for os-brick to new release 4.0.5","message":"update constraint for os-brick to new release 4.0.5\n\nmeta: version: 4.0.5\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: whoami-rajat \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: whoami-rajat \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: I04061d8ab41d4a0ba28103eb418385a753c4aea9\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id065184c67b5393cc3cadca0267f5449dc6e0b89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/465d12d1ac30987a981609407d84a94caf3b519a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/465d12d1ac30987a981609407d84a94caf3b519a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
