)]}'
{"id":"openstack%2Frequirements~833642","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ib08d4d28bb77bc27c9dd318f602f638d3f90a381","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ib08d4d28bb77bc27c9dd318f602f638d3f90a381","subject":"update constraint for oslo.cache to new release 2.6.3","status":"MERGED","created":"2022-03-14 14:54:43.000000000","updated":"2022-03-15 00:12:16.000000000","submitted":"2022-03-15 00:12:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833642-new-release","meta_rev_id":"b143219990cbaf8766cb1543cb665a901f640ac2","_number":833642,"virtual_id_number":833642,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 00:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-03-14 21:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-03-14 14:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-03-14 21:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 14:57:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-03-14 16:11:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-14 16:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 21:25:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8709505536a4403fd26a2945971f8aa67e59c4d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-14 14:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e20e7fecb3aa5eb6e4dd895152b306655daf2fa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-14 14:57:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"efb2891ef356c764e37d850c93f2b0c20513b24c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:11:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ace37e8f54af456c8e1503d23aef16b5 : SUCCESS in 23m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1ad24dcaed47dac07422456bff0227ace6049058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d6bebca4164396b0883985fbd560a6 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3df74635e943457ab3b133db38add488 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bd3152191484e4d9fba0f173790b95f : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e942a197ad4461eb8b5b196eb0827ad : SUCCESS in 6m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/64f6f774c31946dda126d02255dbfbfc : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a07a91be2e443a695ca878b6960a5d8 : SUCCESS in 11m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8f2d01b4076404b9da37084f5a2db1e : SUCCESS in 11m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e221a0fb066e45caae60d593241ba56a : SUCCESS in 11m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ce4d17c110b046b99324bfc6a22cec53 : SUCCESS in 8m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8c1eda778bcd4339ae952992a7a4b31b : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/553468bb2b6d477493962aac8ff85849 : SUCCESS in 11m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ab962aba8d748d58c47bec0c933078d : SUCCESS in 12m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/52a805ed7cb543ba8875c3129103cbc5 : SUCCESS in 18m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/505e4db00ac640c5aa39abc6431cdebb : SUCCESS in 24m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a07beded40c486aa54ee0a1f0c5a738 : SUCCESS in 14m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1bbf5b764f2b4fed972e9aeeee4b6e52 : SUCCESS in 6m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6856df2b35d8425eba0d5a2315ca2157 : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/298715e8a11249bca9a7d7e444e74dbe : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5a929c6814e49d8921726acd41ea619 : SUCCESS in 8m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/19d1c3f29dd240cf810424e0af757f3c : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74533f3e00fa42e1a75ee3bd97ec5e19 : SUCCESS in 1h 00m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d48007b695a41159e856399170a5223 : SUCCESS in 1h 35m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ff22a7a338f5a55a10bc49e8bb3f0f9ff89e9d65","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-14 21:25:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"afc3baa271cbd67d8991af651cfd8404f37ba5d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 21:26:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c3aac493a86263aafb712d890ed30b601e12cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 00:12:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbfa0980fead43de882874a3647eaf73 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b27eb9505ed94f27943488da3b18f310 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bc9fadeb7d047e1a4f202edeb596295 : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7aaba0fca9749d5a11629fcd208ca20 : SUCCESS in 5m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4b1aa8edb214ac7b462b6063913b464 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de9dfdc571174f9fb58c374a81706e23 : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c052b3ff9bfb49909e03c3d788907eed : SUCCESS in 10m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dc2b11abad3b4cefa046ae4f46ddb329 : SUCCESS in 11m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4393a0f3b84d4b40a212bcaaf5651484 : SUCCESS in 8m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/992bfd354cab4924be0fe289870aa940 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8ad339416b24793a3d273c04f2975cb : SUCCESS in 11m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/73d455bd0ca6458494f371dac0f75857 : SUCCESS in 8m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a5b214eb8e9640ce926b42b391f30512 : SUCCESS in 15m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e9a53d89b11745b799462c27abd75fa6 : SUCCESS in 22m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/78bca646865f486784a85fdcec439e93 : SUCCESS in 15m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02a06142ea2842d1970dd8ecab506b59 : SUCCESS in 6m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19f2d56f01c64b9bbb2d1b8d149fdc4a : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4fc1ce40ad5d44ddbced93a1905a4a08 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e5af53ea12c4dc0a7b57473c7872c38 : SUCCESS in 8m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b493c38d9b524e0dbefa43b740e16899 : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f8ebf6df89f4c4e93486661b682efb9 : SUCCESS in 52m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaeac2b45f6a452bbdde4deba1974faa : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b143219990cbaf8766cb1543cb665a901f640ac2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 00:12:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec5a9b3157409082651ba1acb848256a81b70000","revisions":{"ec5a9b3157409082651ba1acb848256a81b70000":{"kind":"REWORK","_number":1,"created":"2022-03-14 14:54:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/833642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/833642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/833642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/833642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/833642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/833642/1"}}},"commit":{"parents":[{"commit":"465d12d1ac30987a981609407d84a94caf3b519a","subject":"update constraint for os-brick to new release 4.0.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/465d12d1ac30987a981609407d84a94caf3b519a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-14 14:54:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-14 14:54:41.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.6.3","message":"update constraint for oslo.cache to new release 2.6.3\n\nmeta: version: 2.6.3\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Idee0d11a70ce9142be2bb8d5c711b9c3da75c1e4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib08d4d28bb77bc27c9dd318f602f638d3f90a381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec5a9b3157409082651ba1acb848256a81b70000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec5a9b3157409082651ba1acb848256a81b70000"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
