)]}'
{"id":"openstack%2Frequirements~833646","triplet_id":"openstack%2Frequirements~stable%2Fxena~I0934183bd3de7dea14ac3a447da3749e76bdd414","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I0934183bd3de7dea14ac3a447da3749e76bdd414","subject":"update constraint for oslo.cache to new release 2.8.2","status":"MERGED","created":"2022-03-14 15:04:52.000000000","updated":"2022-03-15 02:33:33.000000000","submitted":"2022-03-15 02:33:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833646-new-release","meta_rev_id":"6bcad98015af9f3839765fa78e2f7e53908e307b","_number":833646,"virtual_id_number":833646,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 02:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-03-15 01:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-03-15 01:07:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 16:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-14 16:28:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 21:25:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-03-15 01:07:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bfc5b275faf9c0cafdd2b7be41bc4afaad7b78c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-14 15:04:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91f9cfc83eaad1f1deaf17ef37198a279cfb0f09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:28:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7713e82e0ba645da8615506eb3128364 : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c0aac9e05cf3217a6bd6912ea578c12fc5edce92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:28:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12dbbbd7d8bc4a869ba8184082eb35e7 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6453b2daee894d18a3f8768096e0db96 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb7f9c3f232e41959c26b5308908dd29 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9d685065b694fbcb9aafeadf9454e73 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4bd11d75fa7422abb73d93a7f62bba5 : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fb74bffaf3b4fe8ae93a8e71b1524bc : SUCCESS in 6m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/02fe266c842449f6a320c3ff7e3cd77c : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d5ad68ee229a4d9ea6ea6bc6d317b2ed : SUCCESS in 12m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/90a3fb47c9ee442288b7be534d935dba : SUCCESS in 8m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ba1d4dbc8dee4162b436dadcdb359655 : FAILURE in 14m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3b4f92b2f062409e9e788dfb2d3e2d0b : SUCCESS in 7m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8dc256a0cf574cc9b2818c4e428c5646 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/739b416da7fb48bbb848f4408773b205 : SUCCESS in 10m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4fed92ec4e64e99b4ee6834d4f1b14d : SUCCESS in 9m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0fa975177e534efeaf743ec64a3d0293 : SUCCESS in 15m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/75846d2d2cb94e0095cad8936e13cbe6 : SUCCESS in 15m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c65142d603f54f6099fed3dc84d5a282 : SUCCESS in 7m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/42a9d1a960464cb4b9a25422d8404d33 : SUCCESS in 28m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20dffb7bd5e74f0494fcac907cc7d812 : SUCCESS in 13m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/903a7f36b08f470bbc8a3db842129cd9 : SUCCESS in 8m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f15ea872eda47e898fb8506d91cd2e9 : SUCCESS in 5m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9e7451f05f14bc6832b1d171fb671c8 : SUCCESS in 19m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/79138afbea28410bb670074aea8be52e : SUCCESS in 7m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6bf38355fcdc4be2816b734f71febf61 : SUCCESS in 7m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/438912d2d8204e3b9195caa5a026d6df : SUCCESS in 9m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b27d4c15f9f14521bd5fadbe5ef89c6d : SUCCESS in 15m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2881bd022ae74ea987bea3701ef4dafc : SUCCESS in 44m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7645f1ea65c4ad69faa2f5f4b39fbe0 : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":1},{"id":"98e02fb8dd6a1c4829035a146671538a4ffbf1a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-14 21:25:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af40ca64323b03df93a13500b0ecf6d6110e0767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 21:43:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d6430898151043d697286f26d44896f3 : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"83ac3a29859e5e1509467af75efb386847f83714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 22:41:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec2b9999958f492185860cc416ffa73d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d297c018d1b740f88a08654d047193c5 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60575cdf71da4a2b9a33b5fdd9ae7f3b : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/058d72a9082948c3829e2d2a543d1d4f : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1195f3f5f3464cb7b646e01db7cfb882 : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17ca53f6f61141789e247aa948b005e9 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/36e51b9f5435462280b29dc5998cf9c9 : SUCCESS in 8m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f92a2b7160c4d7b8ef9f335fbc0be3b : SUCCESS in 10m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fba37e6bf0c649b9aed7c16e4bb042a4 : SUCCESS in 7m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6879d47e8cef429c988d8171ab114253 : SUCCESS in 10m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d7b8cb2339314cd586b7d8c5ce05a86e : SUCCESS in 7m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/82342183aeb54284846ed4b717ab79bb : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2edf182d847e41c78c6a2675f4388fd2 : SUCCESS in 9m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a6989fd60df141f3bbb8782dcf73954f : SUCCESS in 9m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9b06b7b0dddc4605b6aa99f42e5e928b : SUCCESS in 17m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b39437f7b0f645ff96cf9b0309e4ab45 : SUCCESS in 11m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/322d5ef7cc5849b1bdca46c5b2d36b5b : SUCCESS in 6m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7d89530f53ec4fbcbc55cf8e96d3dcf7 : SUCCESS in 21m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/89ff10ccd0694ade86b44b3d2af4484a : SUCCESS in 13m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9113e8e09103458a808cd3bbe6e2e21c : SUCCESS in 7m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/026dbfa85f354c57b2639a69f88ac4d1 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5e218bce4d94f13b78605da355bea21 : SUCCESS in 17m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4e1cd546101948d9bfd707a91386c983 : SUCCESS in 8m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7dd0daa4aba0440e91480f0356d77d61 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a8566b895b54c00b0525f0fec3c9151 : SUCCESS in 8m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5647bdc496c14b0c93033299bb8d4b02 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d041f6d213440ea91cf56f4a2dd6f09 : SUCCESS in 49m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47164d0c5c084a88bafa3ddd76336d4e : SUCCESS in 1h 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1b1d5669a8a266e5df082324bf686df069010105","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-15 01:07:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0627ea6a6ac4d42e61dc812271b41d6af6c33a99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 01:08:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac1a7f79b6a574c551599d48bec7053502f19ab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 02:33:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e81d1ab9a74513b9f5194d4f335671 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc4a3b42718f4ed18cf863460bd6aa61 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/574dc3a702ab4704bc89be9cf45dcc6b : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec5a9fc9c32a48ce8c8ff19a798ee6a3 : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fbcb35e87904a9c8d7b31bf919cd191 : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/60ab3c109cbc4400a55420b41c398676 : SUCCESS in 11m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/77ed0580e3b34567a56d553f057e0896 : SUCCESS in 10m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/749a46dc21ad4825898eb23eee426807 : SUCCESS in 11m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f4063a14dd28495ebbe26f51e521b444 : SUCCESS in 7m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/619733ab00fe4a0db3177437920e6d60 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1988ad62237844a0ab2a663a94fea164 : SUCCESS in 10m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2505927eb8054a4ba305fafdd7da2dcc : SUCCESS in 10m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/27904154912a4305a01da8fd9322ec2f : SUCCESS in 20m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7bdaf67382da4ab7959ca947f5e1ccc7 : SUCCESS in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dccb231e5759499289fbefd2d73f3917 : SUCCESS in 7m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2418b5cf38034ffaaab988b750049d23 : SUCCESS in 24m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/75e1b9c41ad14d9696413847c591fda1 : SUCCESS in 11m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c317a6d0729144fc90667dc5bc4e0a4e : SUCCESS in 7m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99471314ebc749958eb17b14a066f081 : SUCCESS in 5m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a16111f3c9af43a892384182b6398ee3 : SUCCESS in 17m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fb7ceef60d75463083e9786d3c445b8b : SUCCESS in 8m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c50ee3e9459e4a6b8a704b33436c06ac : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84c8fb44a24f486091807cdd11878ef6 : SUCCESS in 9m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1b0fac9163b04341b2b8502369fa9004 : SUCCESS in 14m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7488556fae284b39a5812c92d365acbf : SUCCESS in 36m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5d03b0629e24e548e3a44b35af915bd : SUCCESS in 1h 22m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6bcad98015af9f3839765fa78e2f7e53908e307b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 02:33:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d513fbdd36f319bafeef605ff16c94c7186fd48","revisions":{"4d513fbdd36f319bafeef605ff16c94c7186fd48":{"kind":"REWORK","_number":1,"created":"2022-03-14 15:04:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/833646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/833646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/833646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/833646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/833646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/833646/1"}}},"commit":{"parents":[{"commit":"1c050adf5a2b0077193468841190bf0b95e9cc33","subject":"update constraint for os-brick to new release 5.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c050adf5a2b0077193468841190bf0b95e9cc33"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-14 15:04:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-14 15:04:51.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.8.2","message":"update constraint for oslo.cache to new release 2.8.2\n\nmeta: version: 2.8.2\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ibaab3b5d6051da62eb4989e2f226936291e954e1\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I0934183bd3de7dea14ac3a447da3749e76bdd414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d513fbdd36f319bafeef605ff16c94c7186fd48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d513fbdd36f319bafeef605ff16c94c7186fd48"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
