)]}'
{"id":"openstack%2Frequirements~834128","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ib944b202535e2c7ed4aed16319f66604fcd367bd","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ib944b202535e2c7ed4aed16319f66604fcd367bd","subject":"update constraint for python-heatclient to new release 2.3.1","status":"MERGED","created":"2022-03-17 10:45:15.000000000","updated":"2022-03-17 23:04:06.000000000","submitted":"2022-03-17 23:04:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834128-new-release","meta_rev_id":"abfbe409b26f6eb574f94bfa751cb8670b7dcf88","_number":834128,"virtual_id_number":834128,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-17 23:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-03-17 21:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-03-17 21:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 11:08:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-17 12:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-17 21:28:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b7b43115185765bc10037f34895dca33b50aa109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-17 10:45:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690f9a3d244ab6a0f20a5be40defc087ab1fc141","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:08:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/55c9bd6c959c484984dd9383f8cfc5cd : FAILURE in 16m 43s","accounts_in_message":[],"_revision_number":1},{"id":"edf2be9c71e6f4f8881011135a7ddfcadb3a42cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 12:06:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98dcb5ead12e4764ac0188be63297516 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8e13a90613f4c9698be7f56f9968382 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5054f945c883440a8870a79296368e3f : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e7077b31ba3480588766d9004ecbbf8 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fe5080e403a49e79cccb1c12cbedf8a : SUCCESS in 6m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cfb118f9fa64856a93d7c9262644053 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/948a5bdf6b414bc6a3fb40531abe85bd : SUCCESS in 11m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0bf82fe282bc45a7bef8815a514c006e : SUCCESS in 10m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f08ce014542740e89fd68805fe21fafd : SUCCESS in 11m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/15a820f54eac44568bd85ffe32ff9296 : SUCCESS in 8m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a61c3902c1514ed3a780edb58415a00d : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/637dcb6cb71e4271a9661e3fe9808b67 : SUCCESS in 9m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/32ada13ec32d47f98d1c5f9c5dbb8de6 : SUCCESS in 9m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bf8ed907bb8f48b9b05b9a14604f546c : SUCCESS in 20m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1d0250ba3edc4626a210bbe0690d14bc : SUCCESS in 6m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/917aee29932248fab281a516d8c2c099 : SUCCESS in 29m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8cefe632dd69440494fca9705f69602f : SUCCESS in 13m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/38615d80894f46c287c31cf5a8f6125d : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e4547e496aa4d7cb6e875356feb3be3 : SUCCESS in 19m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e7c2bef8f7244a5a5ab3208f25eeaba : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/280ed920b12d4022aac3e001852e8031 : SUCCESS in 8m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0b5be372a76741d1a2657103c0caa8fa : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afe51c46388341bd95474a6b5217edf3 : SUCCESS in 58m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35b573e1ae5a4107b66cddd208e4845e : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bb64be6815bf452a8d20b99009f6be2d89404920","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-17 21:28:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d3707cea44c9e2e836e1736f296b3758a3959b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 21:33:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e56bf1af06875a6ddbfbbba660f87593e91c9dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 23:04:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a56b4ba906046c5aff69ab1b6000082 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ad2bf48f79f440fb91b1c6058cc6c07 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ad117794ba4468ca039850484a8fb90 : SUCCESS in 7m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d8ab7bc56114764bae7e7412813ed0b : SUCCESS in 6m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a54f7baddc846a2b2da8d58d440bad0 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e014eeebd9e5432791f1ab2f3c9d8bdc : SUCCESS in 14m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e6e31b589244257abadc7a94df232de : SUCCESS in 14m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e0ecb0be13e45e7bb3ab64ec5b7083d : SUCCESS in 12m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/49f8be08df8e4be1812be3eae890c7ad : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ac2af56faf47447a8fd1ecab13063198 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72b3d800f4684357ac399de889253ca7 : SUCCESS in 14m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0e6481e814ce4eef978683c815b7116c : SUCCESS in 13m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/86cfcc3d4bee46a5ba934f4f74b7058a : SUCCESS in 14m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3ba09dfbd9ec434dba3c302f578cbc10 : SUCCESS in 7m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a1f5620bf2bc40cd8942bdf02da05f4b : SUCCESS in 27m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f77371c48d0a40f4acb6d73e746ba791 : SUCCESS in 17m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd6cb69d1c99477aa282317067490798 : SUCCESS in 6m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/355e40e236d54e7e905b95124ef7a040 : SUCCESS in 22m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c4325ea4af834b1889122eab4d9adcf7 : SUCCESS in 8m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/105d0b4e21584bdbaff793dcbf0cbc4c : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f6d2ea6049594b62b8871d1faa78b5b5 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c54e47c01214974954c371e0bee7b91 : SUCCESS in 54m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f22853fd7f3142128fda4311c4fe36fd : SUCCESS in 1h 27m 23s","accounts_in_message":[],"_revision_number":1},{"id":"abfbe409b26f6eb574f94bfa751cb8670b7dcf88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 23:04:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0aeaa222668c1f63bd7b95e793b77c84900e9d13","revisions":{"0aeaa222668c1f63bd7b95e793b77c84900e9d13":{"kind":"REWORK","_number":1,"created":"2022-03-17 10:45:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/834128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/834128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/834128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/834128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/834128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/834128/1"}}},"commit":{"parents":[{"commit":"849ca1c9110cdfd902ca54633b296f0a1b731c04","subject":"update constraint for os-brick to new release 4.3.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/849ca1c9110cdfd902ca54633b296f0a1b731c04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-17 10:45:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-17 10:45:14.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 2.3.1","message":"update constraint for python-heatclient to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Commit: rabi \u003cramishra@redhat.com\u003e\nmeta: release:Change-Id: I7d7100b57bab09b09ed06dbf1b434d091bbfd3f3\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+1: Rico Lin \u003cricolin@ricolky.com\u003e\nmeta: release:Code-Review+1: Brendan Shephard \u003cbshephar@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib944b202535e2c7ed4aed16319f66604fcd367bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0aeaa222668c1f63bd7b95e793b77c84900e9d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0aeaa222668c1f63bd7b95e793b77c84900e9d13"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
