)]}'
{"id":"openstack%2Frequirements~834254","triplet_id":"openstack%2Frequirements~master~I364900f62a9c3d8201629ac595bd2e206b0406f5","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I364900f62a9c3d8201629ac595bd2e206b0406f5","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-03-18 02:50:12.000000000","updated":"2022-03-18 14:02:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5a7482f0220220aa7a8e75a135a4005a9885be68","_number":834254,"virtual_id_number":834254,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-18 04:16:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-18 03:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-18 04:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ba17e7257c528e404882d24918586387ece69e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-18 02:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2bd27d9dce6c56cce600a498eeb93c100000d36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 03:03:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c3af5fdcbb184baab147b3261d0e41f7 : FAILURE in 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3fcc5eaa49d9dac8e2a8f9d33527cc710005641b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 04:16:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9432b87b8b724c4a828b19d83b8fded4 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4dba4e48d845cb8ea8cbb9c9670c44 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc79ad5dde3a4039a26147974761e432 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06e51c57cdc3444090dbfb52d1f4f15b : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98c14f54483a423b9c5787632dfe9221 : FAILURE in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e21ec537ca054f959812f861845648c9 : SUCCESS in 5m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/06d35e6ad0464fbd923d926d7acaa5a7 : FAILURE in 5m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9cae5da1b6c46d08d2c43be2cf1090f : FAILURE in 5m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fcc488553520401db2e88b5f71593fd4 : POST_FAILURE in 12m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f1f71a0fab37464aa676e687e320ed4f : SUCCESS in 13m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ea4516cf620b47c5a7af7f08bcaaefc6 : SUCCESS in 9m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8348a60ea7f4430295614ba7ca4bda67 : SUCCESS in 9m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d7485bcf2a5c4b9ebb4f8ca75f8fd787 : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac2ef82714a24fada39163843286a34f : FAILURE in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/04dfa094bd87444dbe557781ba2a03f6 : SUCCESS in 10m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ff4797bd85e2449984f755f44071e3f5 : SUCCESS in 12m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1514e976b1404bbeb9839efb97f69bba : SUCCESS in 7m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f23e1fbfbad04e6f9412e32b36542833 : SUCCESS in 10m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3371032748884793ac37919212e1dc9e : SUCCESS in 8m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/10e552d7efd84a8da019eb32c687d371 : SUCCESS in 31m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2756b25f42c14a899472b391b22a7296 : SUCCESS in 13m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ccc6a2c163e74ce9b127eccaa9925a37 : SUCCESS in 8m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b06fef66603647408774710ef374c54b : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7be34e97b16e4e35931424f83a5754b8 : SUCCESS in 24m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8faf0f67c6534d12a8f81808fa9f26fc : SUCCESS in 8m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b0bcb964923c4a8b84f07f898705de73 : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4739563d484047c5adb77201c11b2eb9 : SUCCESS in 10m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8213698c58e74dbaa28057e70f6c93a1 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68752c9a1e3f43f5973455b610ae5b79 : SUCCESS in 44m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d3571e3ea0b4370832aa47968da71f4 : FAILURE in 1h 22m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5a7482f0220220aa7a8e75a135a4005a9885be68","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-18 14:02:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77979a7dc2c7414e0ce0e136c9a9030120b5d00b","revisions":{"77979a7dc2c7414e0ce0e136c9a9030120b5d00b":{"kind":"REWORK","_number":1,"created":"2022-03-18 02:50:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/834254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/834254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/834254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/834254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/834254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/834254/1"}}},"commit":{"parents":[{"commit":"2f2467ad1508f4b125fc3f15c9454dc711baca01","subject":"Update constraint for diskimagebuilder to 3.20.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2467ad1508f4b125fc3f15c9454dc711baca01"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-18 02:50:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-18 02:50:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I364900f62a9c3d8201629ac595bd2e206b0406f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77979a7dc2c7414e0ce0e136c9a9030120b5d00b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77979a7dc2c7414e0ce0e136c9a9030120b5d00b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
