)]}'
{"id":"openstack%2Frequirements~835647","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ib5e264a4f573710bbe523b1b51f5d12f7c1f12ef","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"Ib5e264a4f573710bbe523b1b51f5d12f7c1f12ef","subject":"update constraint for horizon to new release 22.1.0","status":"MERGED","created":"2022-03-29 15:20:40.000000000","updated":"2022-03-29 21:55:39.000000000","submitted":"2022-03-29 21:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835647-new-release","meta_rev_id":"8c706aac2ee17f7e12b2f2ed6d05ad1fa195798b","_number":835647,"virtual_id_number":835647,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-29 21:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-03-29 16:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-03-29 16:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-03-29 16:36:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 15:39:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-29 16:30:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-03-29 16:32:38.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-03-29 16:33:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8937b1036070d6a9d0f3028d5443d24a03e79341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-29 15:20:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"012434cbdcd94982c894113d3123974c94636b5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 15:39:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0ead59fd099947da852e12fc9c0cebdd : FAILURE in 16m 43s","accounts_in_message":[],"_revision_number":1},{"id":"eff10b50b21f0b4667050d425f33a214dac4da68","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-29 16:30:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1eeac1a3a51f403718d9d9b5628b96ecff6f1d1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-03-29 16:32:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ce0ef3dba52cd0b3c339271e8849915287564d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cefb686655948e8b3d33002eef42eb7 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4db4c7163ded4898b485695e891a222e : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c579eec90f3c483385a04f58b29aeeeb : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b41c8241931041b3ad0a1cf03fcf9cc9 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59b991425bdb4121bc7eb833f38a5508 : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd54a0ee578142aa888ab278bed57740 : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06ee8bbbad6741c494c5cb2e1b70554d : SUCCESS in 10m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a201ffd0628e4731bc9bed3931cae349 : SUCCESS in 6m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9b12dcbdadea4eb79c2a2c6dc26ead13 : SUCCESS in 12m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9ac0e91ab5cd4a3d9b185a11a1f63548 : SUCCESS in 9m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6e5cf7b5d00e43b99cb2e2a10c1d6b20 : SUCCESS in 9m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/008c6c390dbb44c2b4a4647aacb0a37e : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1298a6209c5b400bb404f6687f350dba : SUCCESS in 11m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/432960709ba14680a4056fce75e7a918 : SUCCESS in 11m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e1cfb6dc6f28476ca0594187f6477d51 : SUCCESS in 16m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/859514f08de24323a88efaffadc6143c : SUCCESS in 8m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7bac4d5e1f294ba5866047c9d87f9e16 : SUCCESS in 10m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dfea7aaa65e8485ca81f89ca71ad0712 : SUCCESS in 8m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7e0948ffe2874d04b5aa754f0d5fb257 : SUCCESS in 24m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/81e0c2da4ccf498bb860dad0f5dbad9f : SUCCESS in 14m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c89f4932be4541ef80e867cd040b93a4 : SUCCESS in 7m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6a188155ad6c40568ec41c54b8ffc805 : SUCCESS in 7m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0bb73fb40b14dd8a1babbc3580a6d3a : SUCCESS in 25m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1da4df709f2d4395b89e4b5a1449c5d6 : SUCCESS in 8m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5c3c8a051c7c429d8bc060f45387c9f2 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/34816946065d4717ba03b75d493ea6b9 : SUCCESS in 11m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cf213e9b416c478c95ab60e1f0964913 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a8aa7a5ba6d4c8eb36ce4352bdba413 : SUCCESS in 49m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f7f249d903544358ff68708be814952 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1},{"id":"221bf9774953ea4fdfb83c27230fadfb34277bb9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-29 16:36:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac1367c84aba03d642e7fd1ce86b738e64360b0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:37:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"769be923dd614be9007103d64a280548d20422b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 21:55:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcece347bc1d425b8050c884073e270f : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03064080fc1446a9bcd3d8900503b717 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/745a50ce22a0447bac3be901552f4330 : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7df4a1bb32bd4a3f9a785b1a1397646d : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4d93199ce1d4089a392dac2696e8575 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/121492383b114a43bac6f007583e3881 : SUCCESS in 11m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b1b68387f2bc4a24acfd67c1e337f566 : SUCCESS in 11m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/41d47431fce14b109e98827c3db5c7a6 : SUCCESS in 13m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/852bbfdeebbb4619bbc70c52c382d357 : SUCCESS in 6m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/10734001a8484754b136aafcf5f4037b : SUCCESS in 7m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/534d77173d5e4fd490f7cc970e06601a : SUCCESS in 12m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82b7c90a76e5453aa390ced9f04d28cc : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4a7d8fb7284144e78b468278803aaf11 : SUCCESS in 11m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3298b43938aa45ddb9c73a59f31da4c5 : SUCCESS in 13m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a02208382275443390113edd77e026e7 : SUCCESS in 6m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/45db1b55b83b4f7f86e451a9bc2297a0 : SUCCESS in 12m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/28a5872066d341dea2a88c53866e1962 : SUCCESS in 8m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2ee82144bd7b4d018471f39d09f7cead : SUCCESS in 23m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ab03634e94da4780a2bb3d48a9d08b03 : SUCCESS in 17m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b49a6d482682410185f1231ef913038a : SUCCESS in 7m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4055e31328f94dd4b634b6e1c1cfe4b7 : SUCCESS in 6m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bed981909d154744869e65fd1529e09c : SUCCESS in 24m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2720e42b0e174b08bc337fbda715ed37 : SUCCESS in 10m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/87a6ef0b319a4f77acd149e36564cece : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8108fd2d020f4cc5ab48f561b920f620 : SUCCESS in 9m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/49bb94428bf945ec89b60a3074cfceee : SUCCESS in 15m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0eed2122f3d449a88b2475739277add5 : SUCCESS in 55m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da4a9eab3cd542d2adac22344bc07df5 : SUCCESS in 1h 53m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8c706aac2ee17f7e12b2f2ed6d05ad1fa195798b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 21:55:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0ac99fec0eaad1302580aec9de90f4e984b2035","revisions":{"d0ac99fec0eaad1302580aec9de90f4e984b2035":{"kind":"REWORK","_number":1,"created":"2022-03-29 15:20:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/835647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/835647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/835647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/835647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/835647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/835647/1"}}},"commit":{"parents":[{"commit":"0594d9843e670ff9a7df91863a6ba277f85070fe","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0594d9843e670ff9a7df91863a6ba277f85070fe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 15:20:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 15:20:39.000000000","tz":0},"subject":"update constraint for horizon to new release 22.1.0","message":"update constraint for horizon to new release 22.1.0\n\nmeta: version: 22.1.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I1b67d5b258bc0642f4d75db0e6db29bab38a0cec\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib5e264a4f573710bbe523b1b51f5d12f7c1f12ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0ac99fec0eaad1302580aec9de90f4e984b2035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0ac99fec0eaad1302580aec9de90f4e984b2035"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
