)]}'
{"id":"openstack%2Frequirements~835671","triplet_id":"openstack%2Frequirements~stable%2Fxena~I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59","project":"openstack/requirements","branch":"stable/xena","topic":"add-suds-community-stable/xena","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-04-05 22:07:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59","subject":"Add suds-community","status":"MERGED","created":"2022-03-29 16:02:58.000000000","updated":"2022-04-05 22:07:02.000000000","submitted":"2022-04-05 22:07:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835671-add-suds-community-stable/xena","cherry_pick_of_change":813664,"cherry_pick_of_patch_set":3,"meta_rev_id":"1ecd58d79048ba90af1301617badc5c02ea349df","_number":835671,"virtual_id_number":835671,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 22:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-03-31 21:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-05 13:54:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 16:02:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"},"state":"CC"},{"updated":"2022-03-29 16:06:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-03-29 16:52:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-29 18:04:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 21:31:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"784930b8124da39ecbe6aabccacef3f545333070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-03-29 16:02:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0c878ace2897c89233d6c0c8bbc8397c03f6ff01","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-03-29 16:07:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3ff79b709161774f269c7a13affc12e89209721","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:52:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b335fdda7d8b458a883091c289416a1e : FAILURE in 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e5a156deac0c143afc97be392b5bf51750c72d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 18:04:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c9c0411f4c4172a53b5f3cd0fec118 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d32d862c20ef4b1980cecbd8fc276ac9 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7ca13ba62e54990a20743b9e2f32897 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b95a18e918f457f99bd2f552098f828 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f1c4d4a2cb741f5a7220c02bd104f93 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccd7945a654344838c9325ea517ebadd : SUCCESS in 5m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3222bb1d1da9412a8e124759ac236e2b : SUCCESS in 10m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/09d26fe54c93431392fc99326c01faaa : SUCCESS in 10m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f00ddbcb5c3e469190a6425e08829a2c : SUCCESS in 7m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/503de6458add4133a9d962bc678f5efd : SUCCESS in 11m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c0781079e0d748e6ade674e531e0ab01 : SUCCESS in 7m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/95d318a165ee4ad58b6c4d0c01feb2ab : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e55e44a92ce413589735f71dc7e81aa : SUCCESS in 10m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ea421a2861c14a00a6e52d1c552f2d74 : SUCCESS in 10m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b999b65e0c444f3780e1a0e7d04125a7 : SUCCESS in 16m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7dedad8f8e55498fbcfa3e2965a1a0dc : SUCCESS in 11m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dbec631545f3406fb733877196ba6fc4 : SUCCESS in 6m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2bfda1166f3c40d78d198697be8c1fdc : SUCCESS in 26m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/97a24e4082aa42c0b6ca2d31f155635e : SUCCESS in 14m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/14813d9318f64783b02bca5fc5f753ed : SUCCESS in 7m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/338c373b29444f1d85914974da8d334a : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/049d6b4d56884eb19f2cf2c84ccfc48b : SUCCESS in 22m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/60a31eb1f22a4dea81b32466e20a3e8d : SUCCESS in 7m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2bad7d248b4b4064aa50179ad08c0193 : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ffe24f6ca8b343a8abd1eaa4eddcf283 : SUCCESS in 9m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/89a71571cc964d2bbbbb8e75bef3d36b : SUCCESS in 15m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c108991a44584e50a1b7fa62fdc2fd2a : SUCCESS in 53m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/908f062245ed4aa9a33336b697451443 : SUCCESS in 1h 34m 10s","accounts_in_message":[],"_revision_number":1},{"id":"dcbb73244a7b429eda1a67bb42b8e5e54f011737","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-03-31 21:31:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9ff063e15bda12e3131c97eddb620dd794b3d6f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 13:54:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d3c4204aa32d7f4701bbd503c0e39b78dd7a78a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:56:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dda19b4c57971555d57eb74847ee6874e1dce256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:32:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84aefbc62cb44fab8dc15543e73bf961 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/622fbbae9de64c21b1e114130aca3e29 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f41336caed14cab82a271b1d8cfeec8 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb5006e7ccdb46e692886495a4a8477c : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81c8d8596f94413cb0d64215aa91636d : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b2ab9fda8b0416f9a080a53c90b5eb2 : FAILURE in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3f356b990fd4cb5af8c588c3d9256ec : FAILURE in 5m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6acd784259494c878487ec77c7607e2e : FAILURE in 6m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ba0b7b1105484cd1862a19edb9a5bf14 : FAILURE in 5m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/51a0f58ef9df442daf02dc9423c2feee : FAILURE in 4m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73656b75b8694a3b9f104bfe2e4adaa1 : FAILURE in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bcb850b72bba4f70aa8640428178d27a : FAILURE in 4m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/075cd8e2971741ca923f2c0f75dde84f : FAILURE in 6m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9849203e0bfc42c4b8309875a42fc478 : FAILURE in 4m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e13a5711793b4cd59d2fcec667b1401a : FAILURE in 5m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b544ba8a1dc34defb91534dbe1392ba8 : FAILURE in 5m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/97143fac857f4634896526be3d36ac42 : FAILURE in 5m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/56654d61935e4e04a87f44e4183b6a83 : FAILURE in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abae44f5fdc4423ab17f7aa05f495eb3 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9329eda1100b40e7b3fa90ca13ec0128 : FAILURE in 6m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f0244aa108c142c9b43d716b3b6f00ea : FAILURE in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/03bf2f759e3c497eba37b85e026323b5 : FAILURE in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a64574cd56764529934376ebdfb2f094 : FAILURE in 5m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f0aee807d9ef4a83b7132dd70a1f9dba : FAILURE in 4m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f00d3d63efa24d91b5f9096c373cb9d3 : POST_FAILURE in 12m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a83095d161e4dc29b061d5374912cb0 : FAILURE in 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"53d1ba966037df0e30d6d5e433864e8f6dc69f39","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 17:38:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5480570088c3ace3dfaf7918b4064c2f4370f2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:57:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/14cde6176cc54a9f81d43235fdf4ac14 : FAILURE in 15m 44s","accounts_in_message":[],"_revision_number":1},{"id":"71e0c0fdbc35e0f4025fdbb3686c3849697a0ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 19:06:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ddbae9e9b248a5a6a4ec4cae3a4d52 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2debae27204beaa81ada05ea60d357 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b81be4a8c3214129a4be2bea6b1c4417 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/525ce05f2ce949748459f9a97ef08269 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49ec6b8a42574acb863e70549162f146 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2537cc20bc4843e7b37e8a09d5b644a2 : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bcdd0f743c1f49e4a652828a36008d1a : SUCCESS in 9m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c9169600f474073899c2d49981444dc : SUCCESS in 8m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/aebb1bfae718470d984d6717706edeeb : SUCCESS in 6m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b1dc7ce568c54133a044840970fc5fc0 : SUCCESS in 10m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0865bfe2888848189098b0ece5ea4c50 : SUCCESS in 5m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3898a2d6f4bf454c9a4606b080ce3691 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e65cc88bb5af4716863189ed9ed0856b : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/52f58924268e4d1da6ed70e6afc055db : SUCCESS in 8m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f6c14f142e05467f92dac83fcb274ce6 : SUCCESS in 13m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c1fcf0a246114ed4b81def64876efda2 : FAILURE in 18m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0d5d730a97704e3086809b4dcc5c4cbb : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1aec56d24fa24a09877d3d2451a6cd79 : SUCCESS in 20m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7d84d115dc614fb78edd710d5e66f5d8 : SUCCESS in 15m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e823d02ea6ba4e709500fa6e7fdf95a4 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19cd2db21dd04e5ba8809705ec0d50fc : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6965557bff414e458ee4676fe5027e06 : SUCCESS in 18m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/41804147d16a4848bb08800cd83b92f0 : SUCCESS in 7m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d70551096cd34d36a7fd3ae85714d6d4 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/785d16aa616c4c9d8e8e9f5bb285b180 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/29f68b435b8c4eabb8a625a289cf958f : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f565e17272c457682f065bfef7ac08f : SUCCESS in 44m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e47ab5112654ccab5131c73a5fce157 : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c8da16c32f35e57f8452f48b337d9017b3cf9fb1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 19:07:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea9d845848f6ef159c072441381727959cfbf8a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 20:31:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e7c9b37431604e1d87cf99ae56498d38 : FAILURE in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b4f9ed13e35a88872ebaff892654db0dfd8d08d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 20:43:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857abc08f7ea47e9ab0393949abb09c6 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7d9fb69decb4eb79cd69d33811597f0 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e83eadeebe054a1d984216aadd8446ef : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e409d6b512554488acdfc8afc4626f4e : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2412af4cbce4fce8037bb38c1bb16f8 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88d54aa30a164cf1a78d30f69c0f85fb : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/74fb886686f94805879d270f4d050a1c : SUCCESS in 9m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1ecb0ca613c41149113c62723f3f4ab : SUCCESS in 8m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/115ddbc2356249a2983047e255c78d85 : SUCCESS in 5m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bb9ba9ec74d04df3bf1497d1eff09bcb : SUCCESS in 9m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/23f4924beca74860886daa4dcad63f54 : SUCCESS in 7m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e0b2bef03899487f9fc5e1ba0530850c : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1bba19e7212c46749fd3c01661beb421 : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c599273cf3e54ab5b1c78aa489ec916a : SUCCESS in 8m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/564fbd448b7440018f58013496cf3b86 : SUCCESS in 14m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/005b9885f43846a29e56c6d24222073e : SUCCESS in 15m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7d9c1918169f4919b95e7faca4e25352 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5a3d2b07ff9641b18f8ab96ed51ad599 : SUCCESS in 20m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/163038bb8c2449d6841856359f394a8f : SUCCESS in 13m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9329ccaaf13b465e8219d3385b10064f : SUCCESS in 6m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6eba2e4e09824b63a094a40327885e5b : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c13444057b654b07a6686101ec01bbfb : SUCCESS in 16m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9b19a6cb219f43a682416f7a835ca069 : SUCCESS in 5m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f59f2202a1bd4247a5f6fd97f79c2b20 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4beafc6ec25a4324b5509bd3a2966997 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/09165d2517a34ff7a751b33c3c7ffb89 : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11071c034a0c49b5ae1c392c97e350e5 : SUCCESS in 49m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5791b06d7b634d708488c75e900c12ac : SUCCESS in 1h 33m 53s","accounts_in_message":[],"_revision_number":1},{"id":"31d5c91a4a6f83966c795c51a502c2633ad537ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 20:44:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af2e9fde94331c3514081d7c96ea95b0a670c37f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 22:07:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a99a3aa6264a43b93928779a1a0185 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df50143f977640beb62d1e1a5942d4b4 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d82cba0e1dc497482072529eda9ac81 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b898c599c52e49b2b16787cb935400e2 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1055e80ae3fd440ba35d62986b57f7ac : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/acd7a29aceb04d729437ed062ae32e71 : SUCCESS in 9m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8348b8cecfb34ccc940ee37219965aed : SUCCESS in 8m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/17c819bf5808456182eb77e489f753a3 : SUCCESS in 10m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b39b525f3d7a468087c4d74545ba3686 : SUCCESS in 5m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7fbc3153a7ca43f88f93f4e927f773f5 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6c6a4e2f9454f48bc66a341839bddbe : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8045c36ad80b48acb7cdc4c40dff44a3 : SUCCESS in 8m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0ac5c475217d48799c3a619cefd5e954 : SUCCESS in 15m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3e964193c5cb43a8bbc540c255fecf82 : SUCCESS in 9m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bc2f34dbea2748828debf2a7bd1542d3 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2467872bde174838851127aa55318a82 : SUCCESS in 29m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d727807fbf734b4b9af4f20678f32a1b : SUCCESS in 11m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f67a06b618204afe8d41e73fda85ac5f : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/07a9a60ebfa24584a7a81cc9d16c6e3b : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b73a4f344150407fb1ae7dd8c3b3f2b2 : SUCCESS in 19m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dd94c60e6b69403cac743b5d7afaf148 : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2c3089f8b262473fae90cc4fac841fea : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c549a9ec5ef5419f99ddd456e869bca9 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/50d94e8eacfe45e280f19619a56262fe : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94205c3090ba454db2ccb602226f5f76 : SUCCESS in 49m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b60b5d0ec304ed0a05569a59fb0d202 : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1ecd58d79048ba90af1301617badc5c02ea349df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 22:07:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18f957db0e2ac0c27f5642faf465685d9961a6b4","revisions":{"18f957db0e2ac0c27f5642faf465685d9961a6b4":{"kind":"REWORK","_number":1,"created":"2022-03-29 16:02:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/835671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/835671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/835671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/835671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/835671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/835671/1"}}},"commit":{"parents":[{"commit":"a4899ad4ecbe6122e2b2853ea51f9f25218ec77c","subject":"Switch pypowervm to Wallaby version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4899ad4ecbe6122e2b2853ea51f9f25218ec77c"}]}],"author":{"name":"Gunju Kim","email":"gjkim042@gmail.com","date":"2021-10-12 15:12:08.000000000","tz":540},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-03-29 16:02:58.000000000","tz":0},"subject":"Add suds-community","message":"Add suds-community\n\nsuds-community will replace suds-jurko after we remove all dependencies\nto the suds-jurko.\n\nChange-Id: I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit f567f9763a91bcdef9ea94b4a1fab8473d3cab2a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18f957db0e2ac0c27f5642faf465685d9961a6b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18f957db0e2ac0c27f5642faf465685d9961a6b4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
