)]}'
{"id":"openstack%2Frequirements~836221","triplet_id":"openstack%2Frequirements~master~I01f83deb02957632567958ee96c363b84d3fdd9c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-04-05 20:38:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I01f83deb02957632567958ee96c363b84d3fdd9c","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-04-02 02:41:44.000000000","updated":"2022-04-05 23:40:44.000000000","submitted":"2022-04-05 23:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836221-openstack/requirements/constraints/noclob","meta_rev_id":"e67b03b57f57de635bba202f7178b5f267e3c792","_number":836221,"virtual_id_number":836221,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 23:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-05 17:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-05 17:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-02 02:58:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-02 04:14:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 14:11:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5fd03c6b04fbf75955ce4e346729b447fd036cac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-02 02:41:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc6ebf03f1df05c3835c4c242f5513dd60cb68ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 02:58:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d2893902ba9a494087d6c24e61e0282f : FAILURE in 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ebd9ef78c3706e77c9a80feb9c97c2dea2e91390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 04:14:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebe28ba5393450cbf6d8400e2835077 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b309d9cce79547b2a9087c79bbcf7acf : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a5dab247e874e9f82279c0890ce4ef2 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/224fd109843e472d8ddae001cf7017bb : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d853a012eb2403197d55b48cf274f75 : FAILURE in 6m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5606e711547c47d795dc6add3ec9ed53 : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18e667caa83e44c8bd956b5483d53bbd : FAILURE in 7m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3dbb79f4a70648caa0b65108d77bbf08 : FAILURE in 7m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/809a3e7a75184cc6a5ee4fd14cc374ee : SUCCESS in 7m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e0caae756a7843af9b0da999a8d4aac4 : SUCCESS in 12m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8d2ffb63eefc4020996c0f83440fb631 : SUCCESS in 7m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a1e9c06cf7654cc3a90b412420496239 : SUCCESS in 7m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a50c5384a375496784b520b46ca91b7a : SUCCESS in 9m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6e1f21117e734f6688e166cdf876ba79 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5e358bcf8864721b820ef6dce1b6abf : FAILURE in 9m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2846426f6cdd44768614c568b3016166 : SUCCESS in 13m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/433be37ec42a4a70b180cc8b3ab532e4 : SUCCESS in 14m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9947adafe7824c04bf14089334732060 : SUCCESS in 7m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/03f595645b8c4d2c88a5af256b89aab5 : SUCCESS in 10m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/70a8436bf47f4929b0a184d9451b7485 : SUCCESS in 8m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6bccad98e05346b787c9e6e94689133f : SUCCESS in 28m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3a20fb30715e42fc86b456d1ea0eb1b0 : SUCCESS in 13m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/700bbadd2a9c4835906908bc9f2ab9e2 : SUCCESS in 8m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e2e9d5f3d6449e093ec16b82770c289 : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b523819f8bf14547876c241329211afe : SUCCESS in 22m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e177b5bfcb71412d89257309faadeb88 : SUCCESS in 9m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6f03f06e98834e76bc8ecb0b38a953e5 : SUCCESS in 7m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cec36fdc94544e6ba0f7bd2fe3376e85 : SUCCESS in 9m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3612fc81cf4940abb270dc904dc15b8e : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ebbc71928a242c1ae7b7026fb024b52 : SUCCESS in 34m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb7c338731644002b250ddf2537dddc7 : FAILURE in 1h 28m 32s","accounts_in_message":[],"_revision_number":1},{"id":"52fdca99489c72d2859fc656b1c6ca8c530f876c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-03 02:42:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55b26debb5e1cce7f050888738994e18405bc399","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 02:56:03.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f33372f626ae4ff09dfd488e1e09b743 : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"7e4468677c976f79ec2a1ee7de7ca5a22b602ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 04:13:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99f2997949db45e99ba1f4132ae768be : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c14174e641b4f38bf4f12ce189e1762 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aaf8083402646c99ac1139586b845e5 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7fb3dc7a3b141bebee79b42018d4b3f : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a69f2c8d3ba48138df0afba9d6945f7 : FAILURE in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/828182db5e464c42860f2d84d4f8b617 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cdef5519c7374d87bac749776711bfc2 : FAILURE in 5m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d50b5dea2b49485b8ec8bb3897885eb3 : FAILURE in 4m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1aefcef968314ee6b0521cd884520deb : SUCCESS in 6m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/402e63b2d96747c6b62cf46075ba17e0 : SUCCESS in 11m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/de091624040545f69bb220786ada3c92 : SUCCESS in 6m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3085a58e09b1453c8d2c74fea64266ad : SUCCESS in 6m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/dade92584da6416cb42e6919b3dae7dc : SUCCESS in 9m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5a9b3dd4c35b468ba65515e2df1f5b63 : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3528097af18d4e11a0e2e36953fc8885 : FAILURE in 7m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8f4a31f2b237420f9379dd61e32fc645 : SUCCESS in 8m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/570408f01a134fc2904cba5a9a721406 : SUCCESS in 11m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a1e46a89352f4d65aecf02bce21f71bf : SUCCESS in 6m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dccc83d1d0314c05a5e379c35d5d0fc2 : SUCCESS in 11m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6faf039f2f3c4880a18160bef245242f : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6b6fc7620c7c47cebdc3b5f6f2e9d3d3 : SUCCESS in 28m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3892df5ef88b4aa996d53ab63441ab6b : SUCCESS in 15m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0b6cf58bec5848d687575303dcdc07ec : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78d58700da664d35a55f731f1ae06461 : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/356ddf21b79c46c8aa0f9bebb06c6f1a : SUCCESS in 20m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/583e79f6fd6a4bd2b38ee58934928c9a : SUCCESS in 7m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a2bb7921273946d48d2be34443091429 : SUCCESS in 7m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c5e389b45274d52bbe592b39f0f297e : SUCCESS in 8m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9355a2ce57504943a0a576a27531a3bb : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dbb1a8b7e12463fb2cf49015651895a : SUCCESS in 49m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69115764ee474bb19928d5bf6d1102ef : FAILURE in 1h 25m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2a232d19e24cb2b8a5704625a1c61699af079ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-04 02:44:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c51e40319dd2850c3c8dca12d5f36ecad7221b98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 03:21:29.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2b08260526f54d07bfd4d09cc837514d : FAILURE in 9m 32s","accounts_in_message":[],"_revision_number":3},{"id":"77aff50e410ac748e1d320212bb0e0ae0339be07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 04:17:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9456ece56d8d4910a441904d481594b2 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a85441c916f045938add5622f3905ba3 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c1bd1e5fd2b49088ff9fadd79960b18 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f41aa61a43d64f1c9e4f62b0101aff9e : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a27ee6f994254f0c9f902711b6371ae3 : FAILURE in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6929bd68c1014b08a794c01b41df3de4 : SUCCESS in 6m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0f1c7c686ec4702a11386d6ba9a5fe7 : FAILURE in 6m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5cf52cde7dc44154ad037f2128bf20b2 : FAILURE in 6m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4108dc6f0f98483682fc338641e956a5 : SUCCESS in 7m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d48060d984414d4dac3e008a05892af3 : SUCCESS in 11m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2368c77db4b44b8cb6c60defceb3c1f9 : SUCCESS in 6m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3e19d3b8620745028f965e4fc16f303c : SUCCESS in 6m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bb343b1287d54d17b755d6ea8b191196 : SUCCESS in 9m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/13f07761c6ec42ecbeceabd3d989d90d : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6817b6f79ee64604b8b4a2ba53b6194e : FAILURE in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c62dd87ddd014b009fdda7513c962c8d : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e7284e3e511440579f48c3dea1ac7d8b : SUCCESS in 13m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e51decfa19894a6794f10cfda89b8392 : SUCCESS in 6m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/29b755c8a0b74f2ea70a5077ddaf87ce : SUCCESS in 10m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4e17ed67d3d34aaeb8b7d396243a12fe : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e652021059bd4461827ac1a0389c88ee : SUCCESS in 25m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b8fb9f9615e646509483e3aba81c08dc : SUCCESS in 13m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d5dcae37c49b4cb5a6ad09d653e06500 : SUCCESS in 7m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a7da83b280d4e7b9b02409b38896f71 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7ebf90b9c5c4aa7aae333e4c0441112 : SUCCESS in 22m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1ed29acb96ff4c7197b8e8024d148029 : SUCCESS in 7m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b942c5be7d0244f6844a5d12ad46f289 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a08a6563420744f3821da356122bde2d : SUCCESS in 7m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b3010f1ef6bb480d8fd848200fafd85a : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6961e1a34b643d99adb2a63c5e90ec7 : SUCCESS in 56m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44fc5865a93f4a08a773b00d52a2f3c3 : FAILURE in 1h 28m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b40291445a37db597a06f14e5361970355ccaa2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-05 02:47:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec2a79f2569724933789f4bbbd282e3322fa86c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 03:03:11.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4a29e261a424446caecf2bf81fc40832 : FAILURE in 12m 53s","accounts_in_message":[],"_revision_number":4},{"id":"d0fb24048fee0ef89114f73e8d67bbadfb5bed8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 04:19:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b304490ecb542bdbacea966dae4a0dc : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82028c10a53415cac9800ade1829e43 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27b9d65bdd3f47329692a2260b0838a5 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/358e3cce0c0843ecb515060bf7050cf9 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9c46a7b51664e62a7b3ec6c4564710e : FAILURE in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a78b093f2c5c420ca719d0968bbbc4a4 : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8af55729d0ad4f31b1dd942610ebf2e5 : FAILURE in 5m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/33ead3392221493d859d2f1e1a3a38f3 : FAILURE in 4m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b69fae403920448a949631bd8f58be0c : SUCCESS in 6m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f91e37f9f66444c2af0a01ebc87b7253 : SUCCESS in 12m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/53a87afcc316450c8d413527ec9b38de : SUCCESS in 5m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/130e2198f9f34f1ca461fa24efe6ee80 : SUCCESS in 5m 48s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fb82c0267d4e4eccb955323b045a8c15 : SUCCESS in 10m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dc3e314655f44bccb8c579f677281d98 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d000e9b3ece458193f0dcc0785a74fe : FAILURE in 10m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d9c7d43f1bf64484991f21a505eda6bf : SUCCESS in 12m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/94704d22247d4d5d9bd8ec83b12b416b : SUCCESS in 14m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ba7fe866ed9a42ca86daa56cf2151f03 : SUCCESS in 7m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0c49f94941484161bed3273e6ca4f775 : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/62166621ff174181a6941fd0ceb8aa4b : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1a2f99c7fdd1469195532bfaedb7e02b : SUCCESS in 24m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c996eff73b0a4ee885aebfaca2f68108 : SUCCESS in 14m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/99fdc95475b4422c814a3206502b746a : SUCCESS in 7m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f009556d42894499be6b6955ec7e6401 : SUCCESS in 5m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f72075748f14da2807476340f4feac8 : SUCCESS in 24m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/06e93e2df3234a728514e36ce07113f8 : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8f221495adb74957a63e314570c306a1 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b01a1d37e3954d92bfe9c26482a45ccc : SUCCESS in 9m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/22589a9cc8ea4eda99ff545b951de490 : TIMED_OUT in 41m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f343c6a880054c8291e600acf0f0651c : SUCCESS in 34m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/824ab5bf8bd648d380cf00bf8d8e497e : FAILURE in 1h 28m 30s","accounts_in_message":[],"_revision_number":4},{"id":"abf4978d8a4b68aff3362bd9586b01d51aa0bb77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 14:11:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6d898282062434247c8437cdb5246098430ec98c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 14:11:44.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":5},{"id":"d9b696895c96b5fd926bfbe1bf2a2d8de2271fc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 14:12:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b3177b96d4f1fbc43bda7e72ccd4c043e075832e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:59:36.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d16cee504a8d4b37959b861bec3508b5 : FAILURE in 18m 48s","accounts_in_message":[],"_revision_number":6},{"id":"d93fe9f96997c8d1acc6d942f6c7dabb0b91f9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 16:08:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2211e592aa43ff969bc9fc75fff27c : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f56e1b2ccef44abaee532474b747b89 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5f95099e3994e54be3754a1c7f5ac7d : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13e31e870bd24534a956638ea0818aa3 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08be4f9ea2b3479aa958c562d4b198e5 : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc6bf8b4b18f411baf84ad60f58735c0 : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2b381b3785334ec3a491f336601b9ae0 : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fecfada254d34e35a33642917e589313 : SUCCESS in 11m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/56810ffe7bc7463280beaa15d278e83e : SUCCESS in 6m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c8af000058204e62bdb411f46d4ae13c : SUCCESS in 11m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b68fc19580c24f1794698b3da87e5dc8 : SUCCESS in 7m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2f40bec9175d4109bb820b77dfa63fd7 : SUCCESS in 7m 10s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/33766f14c3454fe28b4b2e41272332f6 : SUCCESS in 9m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4564895eed4f445fa4bae14094ba0afa : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/acd4bbe9c43e48a8b683be601cb60523 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b23f9d1bbc7e481b83b79d939ea0c066 : SUCCESS in 9m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/968ba84162d64f4b931d83bb518aea9d : SUCCESS in 12m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f3a3118b8b2641aa98d5d39c3092748b : SUCCESS in 6m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8e3c57bb8c2749fdae492318a1e6a0fc : SUCCESS in 11m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fb5ceae6d8de407f99d8b5d5ee24f18d : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3f6809e5eac54c5d8d10da88aec674d8 : SUCCESS in 24m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f2d89a647bd94613844c1fd5228c9652 : SUCCESS in 15m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3642e0ef5dba41298d99cff2f3240526 : SUCCESS in 8m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd759e4a1e2142f28a3d06a9dbc28f26 : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a061c0a37ff14e46b94a38975421b9a6 : SUCCESS in 21m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/615ab1dfdf17499fbb29d49310dbc4d6 : SUCCESS in 8m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/90aeea66211a42feb3adfa32f914497d : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/623c911e850542388fbb495f2f852097 : SUCCESS in 8m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/432a61caf2b340ca98b06afa7f4065ff : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97607b1f9ff3495ba1ff26e1d2109fd7 : SUCCESS in 56m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b127687f9f13417897ecfbc5ea5ce74a : SUCCESS in 1h 34m 16s","accounts_in_message":[],"_revision_number":6},{"id":"3756b1d7c2d6ac9a47e558a50c16a3678dd0a2d0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 17:38:38.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"80590d279370e82f123986b6c1eb4f0613c9b7ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:40:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5e9b3bd6a1614336d137981b51788c8993b41de2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 19:00:31.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dee0be8db12493faf17cb74def33e3c : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b82fcfeb0884f32971a675467c36153 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5934bf71b2984ba691164da32ab8d8d5 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68920b2dad2944d48d2477a08d639fbb : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b1b22638f8c4cee8ebbd73c5cac8c9c : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85cfc40c8e8c4f4ea23178effe6c4570 : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d14d47e98edc414e9a83560a44da52b1 : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7175929f84464766b11eb9cb4b6de47f : SUCCESS in 11m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/46f5065644ed4282a66945b6cbec1700 : SUCCESS in 11m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/292dc1a3efde41599b7b4183e410d55a : SUCCESS in 4m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/085af4921b6d4f53847e423f5d319b9e : SUCCESS in 5m 44s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e1837bc4bb1947509dbff49984614537 : SUCCESS in 7m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4c3a13dca0f64b91bdfab86d45668459 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b723ce3776f4e369a850899b8524b62 : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a58e2dda2534154bdbb22ce33522009 : SUCCESS in 6m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3975c1426bb442969ba5e53f3fbbe676 : SUCCESS in 9m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a3731a033c324c549d2e3ed06f7f8e96 : SUCCESS in 4m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3ba3c44e2a534f5c92455d14e18da0f8 : SUCCESS in 9m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/492d6089201445338ac29c19690339a1 : SUCCESS in 4m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a25ee990a17743ad96b670d080706072 : SUCCESS in 28m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/41fe93a1f90d4eaeaa49e6fa6fd3c925 : SUCCESS in 14m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/34ea0ed8e06b46efbfcfdc913cf3a074 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2ecb291641ae4fd28768861b052c6db9 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da6cfd47a2b04a5ba954c03af50f85f4 : SUCCESS in 17m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/78a7c5cef03e445d9b3b1d47940fccef : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/81450734077447baa948ecee727d424a : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ba0c2db9a604007ba00ca598ddc30aa : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b882b7ea6a1e44aca29650b461b9f93e : TIMED_OUT in 41m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/001f77799a1848a69591da0a722aaec7 : SUCCESS in 33m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc7fcb951b564592853a44dbad07e007 : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":6},{"id":"25b6dda9527b936cdce7d9b413553e0485c80fcf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 19:07:50.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"40fb19d20484f75f3e6ec0776d1d78892c7b047b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 20:20:34.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9ef945160b0d4dc7955fba821170e587 : FAILURE in 18m 32s","accounts_in_message":[],"_revision_number":6},{"id":"948dc803551e7e55c0b685629b96e616f4b1f3c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 20:35:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454afb2ddb134ba09105966d56c365da : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc00dc3cec32465584fc71b12627f655 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/127805d32fe04037838175579f0f5723 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26438280d89a4847a403f49362199feb : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d74a84034474ee98c4ea22825eea2b3 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05cf88f752a6413bb0bb6fdbda5611e9 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c4eec0a1fed4ea1b1123cef5568df01 : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/95017aff066e4c40a7862852c29a8024 : SUCCESS in 11m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fb587778a0914d3a86f9012c4d0d249a : SUCCESS in 6m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f4bbe38b272f4da3a5b4a220a36c3795 : SUCCESS in 10m 50s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8bbeb882a11640bc9164c29f508dd8dd : SUCCESS in 4m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/56dcd35087834b66a13214bcc3b2381d : SUCCESS in 5m 54s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0f6a6e31079e4900858171efe061073a : SUCCESS in 9m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/76427c1c19f6473882b22f01b64f172f : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/757506af555448a18602e794ecca0b12 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/366f9318049f4f2ebf62431a4daddf39 : SUCCESS in 9m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/33cd575a058849b49c9c78dc561b4c5b : SUCCESS in 12m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5a12567854ad4742972a58b3a7feb986 : SUCCESS in 5m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7788360665f9420393422c0a9864f3ae : SUCCESS in 10m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2d57a939cb6745dbb7dcb6295eaab823 : SUCCESS in 5m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2947b4a4d0c14d61945aab4837b894bd : SUCCESS in 21m 24s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6990e284afcf4612b5a65ceb86ffee5c : SUCCESS in 13m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/04c29fc571824bb1b53fe3e041bf67c9 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d07d838efdcc4e579df21c9b39a875cd : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/91fba3b7a12e4d7bb1ef06921b9cdf29 : SUCCESS in 19m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/745ef1b788f842a5b5e75dfacba148d2 : SUCCESS in 6m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/96d695580c2545fe85271cb691a3bcbf : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f31eb2db2534313b2a2b9ab69d9b54c : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cfbcb3e53bb64a47be1576ca7926bfaf : TIMED_OUT in 41m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8b006781b50497ea2019594d966b18e : SUCCESS in 46m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69dd2bbfaeaa43978509309134045361 : SUCCESS in 1h 24m 24s","accounts_in_message":[],"_revision_number":6},{"id":"045d77e836e3fb34b47e1789274d16de45e1d4a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-05 20:38:33.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"fa0a53b296288fb0a4bc367f9e0e9c70bf808a26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 21:03:31.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d0d5cf1231464778b6f516891a50f523 : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":6},{"id":"9a54e71d04bf24d108eb16140ad24acfacf5acf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 22:26:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67cbdca8591c465cbb6c096477dbff70 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64332277df1d4f5f9d12f703832f01e9 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a81652b51344d37aa5471d6c9332742 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92c2a7f5da9f4e6f8d7102f4917ddf2a : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f03711d669643819d816c2b40c74b34 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95ca126075e7487293e3ac42cc194f9f : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f56967cdd5a341ef9b79e87ad78027e2 : SUCCESS in 10m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d09a58e5ee29465ba7179bb0a6f21e3f : SUCCESS in 10m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dca2395e4fe240cd8ea6fb2f3c5397b9 : SUCCESS in 5m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e687da684c164ab0aa8884ef65faa419 : SUCCESS in 11m 53s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9f2a0216d5f24c08bab4b6b9b1841607 : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/abb7599c7ad44a54a92acf6f825a6cc5 : SUCCESS in 6m 05s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b8966ab039f54e82829f529377f777a0 : SUCCESS in 7m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/57bf8d6372464fd0861f5601ea7e3c35 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01a9a9d9f09a449cb7dc18b9bddfb70a : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ac8ed1dcfe6b4e00ad034219ffcad211 : SUCCESS in 9m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ede64483271046959246efa47494b447 : SUCCESS in 10m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/21a0c915c17c46a2838bcf5b7bfe5b29 : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/18992ae3010543d29e2b37c49c1fc5c1 : SUCCESS in 8m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8abf5beb29d54b518bcafa7845814bae : SUCCESS in 4m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9ed5fa48ccd249da915f0703882bc495 : SUCCESS in 25m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/948fe6f5268245c689c4710f44f71f65 : SUCCESS in 12m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2559587424744ab79960e96f282afe22 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/64b1d41e0d4f4387b40c410bc8b6862b : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4f4820bae7a483a9bd67ae84e8d8107 : SUCCESS in 20m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ecabffcf9bb7499d9de6f3f5fe6b7c6a : SUCCESS in 5m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/28ea2958048f445b80c817be9decb53a : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f3868952d3946a0a0f44ac5b7b60ba9 : SUCCESS in 6m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5c7c98dc1a634113b08053b3009fed63 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca81c316a9994e878130b6b02684f051 : SUCCESS in 50m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fe01ba3e6f64d36823c9128f2419263 : SUCCESS in 1h 42m 18s","accounts_in_message":[],"_revision_number":6},{"id":"cca734c7fcf12d011c1979504e36d329797fb4f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 22:27:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1aee83000f0ec6915c103827bfab076994e60998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 23:39:28.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91f5407c3c246a79aad0b8123b871c4 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22a90d4e32841bd95d5076415de311e : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4236384ed7746b8a0c27cb34a480f9e : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc8de2a134b443fc87a49c3c75955d79 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5121af18960a4fa181abdd87d50ec8db : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/949de00133b94a1fa5b7ed33704f3fdd : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f5e8708442644f0bb8643f865a0e0331 : SUCCESS in 8m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99eda3f9bc8241be808e76478029ed52 : SUCCESS in 8m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a1c032e4bbf7431eb3f920f1292c1f52 : SUCCESS in 9m 14s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/63e493c0109a4aaeb8883f462ccb58c9 : SUCCESS in 4m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/82a9a8f948804ccabd233109b3771bae : SUCCESS in 5m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/35c7ffb8727e4ba3bb2d1a452a307693 : SUCCESS in 7m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0e1ebd2dca65455686ee1d28c4536e46 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f32756acca30485cbf940213d2d0e158 : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/94cdf9b7127b44409a36fda827a0fa4b : SUCCESS in 7m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2d60224c9db94f458dc657c1ff772e5a : SUCCESS in 8m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/265ea760f93341ea8db73ff15a3fac22 : SUCCESS in 4m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9756ad18e9f2428b8bdd6d4304d20fc8 : SUCCESS in 8m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2c58ac9683904e24b8ee033febf359e4 : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b203dd8669e54c5da4276044c679c89c : SUCCESS in 20m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/118d85fe66834276b36df9dadd02356e : SUCCESS in 14m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d7f4f99af63a468ab4b88e5592bf59b0 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/447d2565f0b2494a9687871210fc5224 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de4299fca3714908a1fca4cbcb3c876d : SUCCESS in 20m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0cbc11619bdf4d05859b4b167d620750 : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d14f39cce4354916b321669d5b74e7bd : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e137b7cb6694c52947a671e1dd3d2cc : SUCCESS in 6m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be3115f5d30942afa936e7946109cba4 : SUCCESS in 9m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a99fb87719a44788a73233fb1f4e44b7 : SUCCESS in 42m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4a8089fe36541f1827873b8697792b0 : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":6},{"id":"a2243c2f67982e3f1d94e65d5c7e89bb804c325f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 23:39:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e67b03b57f57de635bba202f7178b5f267e3c792","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 23:40:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c587a6ee617e4eb789d0bae0ecedd07b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f4423fb5e06125e13ba91a0c1f7d3b595517bb59","revisions":{"f4afb2347c2f3baee33e2a4b495405857e9457b2":{"kind":"REWORK","_number":1,"created":"2022-04-02 02:41:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/836221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/836221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/836221/1"}}},"commit":{"parents":[{"commit":"0ae5c845fc5c2381c54402cd6b5efe5419dff1c1","subject":"update constraint for tooz to new release 2.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ae5c845fc5c2381c54402cd6b5efe5419dff1c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-02 02:41:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-02 02:41:42.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I01f83deb02957632567958ee96c363b84d3fdd9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4afb2347c2f3baee33e2a4b495405857e9457b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4afb2347c2f3baee33e2a4b495405857e9457b2"}]},"branch":"refs/heads/master"},"4992f9440e5b22f80186bd38509a7ed70d389f57":{"kind":"REWORK","_number":2,"created":"2022-04-03 02:42:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/836221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/836221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/836221/2"}}},"commit":{"parents":[{"commit":"0ae5c845fc5c2381c54402cd6b5efe5419dff1c1","subject":"update constraint for tooz to new release 2.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ae5c845fc5c2381c54402cd6b5efe5419dff1c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-03 02:42:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-03 02:42:14.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I01f83deb02957632567958ee96c363b84d3fdd9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4992f9440e5b22f80186bd38509a7ed70d389f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4992f9440e5b22f80186bd38509a7ed70d389f57"}]},"branch":"refs/heads/master"},"ebb1f2928d9c410dfd106a202dc24f275a2d4740":{"kind":"REWORK","_number":3,"created":"2022-04-04 02:44:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/836221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/836221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/836221/3"}}},"commit":{"parents":[{"commit":"0ae5c845fc5c2381c54402cd6b5efe5419dff1c1","subject":"update constraint for tooz to new release 2.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ae5c845fc5c2381c54402cd6b5efe5419dff1c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-04 02:44:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-04 02:44:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I01f83deb02957632567958ee96c363b84d3fdd9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebb1f2928d9c410dfd106a202dc24f275a2d4740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebb1f2928d9c410dfd106a202dc24f275a2d4740"}]},"branch":"refs/heads/master"},"6104f1013faeb1737f2e40192f47a2f07e0892ba":{"kind":"REWORK","_number":4,"created":"2022-04-05 02:47:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/836221/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/836221/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/836221/4"}}},"commit":{"parents":[{"commit":"ebba5ca1181490597366d79416bc65b49aa4450b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebba5ca1181490597366d79416bc65b49aa4450b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-05 02:47:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-05 02:47:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I01f83deb02957632567958ee96c363b84d3fdd9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6104f1013faeb1737f2e40192f47a2f07e0892ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6104f1013faeb1737f2e40192f47a2f07e0892ba"}]},"branch":"refs/heads/master"},"0a8fe7d9d479fd51b225cc95423d66c0b28260ac":{"kind":"REWORK","_number":5,"created":"2022-04-05 14:11:39.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/21/836221/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/836221/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/836221/5"}}},"commit":{"parents":[{"commit":"ebba5ca1181490597366d79416bc65b49aa4450b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebba5ca1181490597366d79416bc65b49aa4450b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-05 02:47:15.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-04-05 14:11:32.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I01f83deb02957632567958ee96c363b84d3fdd9c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a8fe7d9d479fd51b225cc95423d66c0b28260ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a8fe7d9d479fd51b225cc95423d66c0b28260ac"}]},"branch":"refs/heads/master"},"f4423fb5e06125e13ba91a0c1f7d3b595517bb59":{"kind":"REWORK","_number":6,"created":"2022-04-05 14:12:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/21/836221/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/836221/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/836221/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/836221/6"}}},"commit":{"parents":[{"commit":"ebba5ca1181490597366d79416bc65b49aa4450b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebba5ca1181490597366d79416bc65b49aa4450b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-05 02:47:15.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-04-05 14:12:29.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I01f83deb02957632567958ee96c363b84d3fdd9c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4423fb5e06125e13ba91a0c1f7d3b595517bb59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4423fb5e06125e13ba91a0c1f7d3b595517bb59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
