)]}'
{"id":"openstack%2Frequirements~836324","triplet_id":"openstack%2Frequirements~master~I5baca49822033d63f017d1a200188b4be5e9893a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5baca49822033d63f017d1a200188b4be5e9893a","subject":"update constraint for os-testr to new release 3.0.0","status":"ABANDONED","created":"2022-04-04 12:45:09.000000000","updated":"2022-04-22 17:19:09.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"229c8a4e48fa9c25f7c0a17998bbbd961fd83770","_number":836324,"virtual_id_number":836324,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-04 14:12:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-22 17:08:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-04-04 14:16:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-22 17:08:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 13:10:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-04 14:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 14:16:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-05 09:09:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-04-22 17:08:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-22 17:08:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"d8e10cb7789f054e61eba5baab8b2a747f5a9797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-04 12:45:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52a9c8389a73bd56fcab717eca481011ce3a2809","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:10:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/38e485c833dd4c6bb86b29c62a3b4381 : FAILURE in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2ae9770bf787bfccfa49a837b3c20d57dfc8675c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 14:12:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bca43c49c0e4e9684f79208b3db6671 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c086a72b824b499245e6ac87bf759e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa80e04fa88649faac5ce12949b3b28d : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10625cfffbae47778324420420fce3f2 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a0e2063b5304a1480b009d031dfaf9c : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aac3d56b7f1b4b80a1768dd942255be5 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c1ce48f8aeba49a188fa303783053d26 : FAILURE in 5m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf60b68211aa4eec85d2d17f201974ec : SUCCESS in 13m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/be942a7a81714c23b92731bc72bf3718 : SUCCESS in 5m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/992ab53b1ccd4b9688a17cd8c0e17959 : SUCCESS in 12m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0cffc3feb6994d91aa13fcd9b3103881 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dfc724df601049bb8cd0057593ea3405 : SUCCESS in 4m 48s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bfb7fb13ed8041f29f55e21d8d711acf : SUCCESS in 7m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c122caaf677c44088466ff0bc15c7695 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e09503a346f6490385e036c2e33d77b8 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2bd11770b8ab41489d3f46ed6208b65f : SUCCESS in 8m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9ed1288c3c4149e984a88ad4762805eb : SUCCESS in 11m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/59817d0dcd4e44649e38983a2d7128f4 : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/601a9c14bb2246b48cd659baf0211f33 : SUCCESS in 8m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/de53fff299a244e085b6e0df3fa2ef00 : SUCCESS in 5m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c86da2b25cdb490e82ddf446c3a5421c : SUCCESS in 30m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6219990c64104ad6bcc5e6f599422151 : SUCCESS in 13m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1d14374c065247f5a0235222955436fc : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43d0b03d636d42b08f1eb546d0bbba43 : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/948905c3cd214c629f9c132d2eb178a4 : SUCCESS in 23m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/52ebbfed3c0846649547155a4fed91f0 : SUCCESS in 6m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/45ccf28d6e4e4bef82bafdd4edb0bb06 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/045d1087a5b948d9bd38943c29a46686 : SUCCESS in 8m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/830e9174005e4969907b53d6adaf1d54 : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a74e1e8ed0e04d0da9e955e0436e7aff : SUCCESS in 54m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/615572e744644adbb5ab1b3be10fc67d : SUCCESS in 1h 25m 24s","accounts_in_message":[],"_revision_number":1},{"id":"83c569a39a12559ac7ae44660fa7337b6bd8787a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-04 14:16:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77a3d4953ad3cfbc63afb2494eaed81cf35fb384","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 14:31:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cea001722fc94e86a9d7f5cabbb28b3e : FAILURE in 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4d0aa49c11364054546477badf7856ca50083225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 15:41:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb4e60833d141ad9732a646975b4ec1 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d597201186654518be92cbfeaeedce0c : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e5a6c41c0794e2f8cfa1f65f8b5312d : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af0bef9f4338434f8ad3f2f490596a6a : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fef284f4d204f69bab8166c1223063a : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12bdd66191414b52b14ae501052e8b60 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c503c79e47d043cabc4148eb442d5f75 : FAILURE in 5m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b819a67406734b64a3fbda6b11727b61 : SUCCESS in 9m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/131da2f2d634451bb6cb813ceb3ae705 : SUCCESS in 4m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fce12ae933a44b38b88b2ef807011224 : SUCCESS in 10m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/57baf3573f5f4aacbf03c295375a3a12 : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a2a50258ae4a472d9fc5f43a3ad7478b : SUCCESS in 5m 35s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e94ae4f16a184221bffea239f584cb67 : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9e1e20ca91d24baca1c6e6a0593c616c : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fff6e297ab034564be1df2450ea5e13c : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3bc48ca63c234c5e9b04ad143f801977 : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0df022565a38418983b5febb1c9592e9 : SUCCESS in 13m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/95849b9a03ae4e95bbef1d39f6c7f6b6 : SUCCESS in 4m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1225b64439c445eea623e9b0dd19da4b : SUCCESS in 9m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cbc32fb990aa412d8bd98cdf3d44e557 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7b65bded569b41fdacb80d911f789613 : SUCCESS in 20m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a6019cc8a0ce492d9313bccdc00b0dcf : SUCCESS in 12m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cade174c056345df8b246488c49b1731 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7fac415361fd420bb06e25bf4924fda4 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/faf33b7306f94560b2ed6726c7e715f0 : SUCCESS in 18m 14s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c2ab084e8f9a48a8a241e30ea5d21059 : SUCCESS in 6m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/96124407b7ef4fe39c2152dbbeeca59b : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41cbfd17f6f7436eabaf2f5e1002d58c : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c5edf39382ef4f588baacd862cf9b0c8 : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/060a52bf539c4fb98fa9b1dc9ccab100 : SUCCESS in 42m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7538af703c8d48c0a9ba91c053309a9b : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f9d5b1f12f3317b0d4dea29aafd3624762693b69","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-05 09:09:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"034f9555a093fb9a377801177fb9af931c059c99","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-05 09:09:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa2eb6d450f57bf6a38da11e97eadaa7f83cfb8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:52:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/258ffbb069b644e286ac24952096d6bc : FAILURE in 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bb76c2c675990ad8f6a137a816fcc1476a38f4a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 10:54:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa214445653425aa35d5d25f3d19947 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64388c7620c541b9ad4a1eb5eca82305 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5767b5c03964ac9a0d727b02db55fc8 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab1b359fd59d46abb0f8becff3aa4ae0 : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3af470389f445a0a5cb74d56972baf0 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c1fa12445f64b8196b7febdaf4f5052 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a53fc29d8f14e88b46c6028143c5b42 : FAILURE in 5m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d579da9505134b359efcc9d7b90dd840 : SUCCESS in 11m 32s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/04268b420d53497ba0ab5867b04b2042 : SUCCESS in 6m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/581e3f10afdf452c86c52cb2a70c8606 : SUCCESS in 11m 01s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/794bd6148dac4168916f011c82b2e6dc : SUCCESS in 7m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bf115e205e85425d91abeae818e0298b : SUCCESS in 7m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4093dd328d204be6b00b41bf1f93831e : SUCCESS in 10m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/22200b411f55473593402be4906f0475 : SUCCESS in 13m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2da5989451c74e62ba26f17a1a9cb9d3 : SUCCESS in 10m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/791512d8413f40368a3adfa350956c4c : SUCCESS in 10m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2f13058da21642c79abc396195f4deaf : SUCCESS in 12m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ff57c9a0b9de47b28228f664a05d69b4 : SUCCESS in 6m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5b26fed4541847c194f933dadfc02447 : SUCCESS in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2d669550f5cb4f52b99ff62ff326c2f2 : SUCCESS in 8m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8a93b5fb784c443cb72efd71fee13e54 : SUCCESS in 27m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7621d9b5610449219bc87b152977d35c : SUCCESS in 14m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/68cac0e551d24fbd8239b71d5ce0a5bd : SUCCESS in 6m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/db7b08810778483b82519abd6b4357e7 : SUCCESS in 6m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/771e15652dd249e4b5218a58249edb0e : SUCCESS in 23m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/479c3fdd47844f97b1657ef33ec2cbbe : SUCCESS in 7m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1abcf917b0c04e32b4f16ddb4a03c9c9 : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f3ebbf4db70e409186d98a9af0dec4c3 : SUCCESS in 9m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c6d6cd9300eb4fcdbab46f92368c3973 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c198eee4341472d872197bd5a1064ea : SUCCESS in 55m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5a2d20541b94d5aa9c6224149227c3b : SUCCESS in 1h 26m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2dc5ed4cd447f2a2f713191f40f5d16c079cc297","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-22 17:08:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72f96912d2ecac3267be12ffc483f3e3036f3dcb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-22 17:08:39.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7876712a9161e7bb7f15b78990cfa9c1e7ce1cb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 17:08:41.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 836324,1","accounts_in_message":[],"_revision_number":1},{"id":"f852f45c382e2374e0542f21226de857dcbc9026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 17:09:03.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 836324,1","accounts_in_message":[],"_revision_number":1},{"id":"229c8a4e48fa9c25f7c0a17998bbbd961fd83770","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-22 17:19:09.000000000","message":"Abandoned\n\nwas merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b0e9db3636448d927bb0c371eee9311715f79db","revisions":{"9b0e9db3636448d927bb0c371eee9311715f79db":{"kind":"REWORK","_number":1,"created":"2022-04-04 12:45:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/836324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/836324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/836324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/836324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/836324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/836324/1"}}},"commit":{"parents":[{"commit":"0ae5c845fc5c2381c54402cd6b5efe5419dff1c1","subject":"update constraint for tooz to new release 2.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ae5c845fc5c2381c54402cd6b5efe5419dff1c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-04 12:45:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-04 12:45:07.000000000","tz":0},"subject":"update constraint for os-testr to new release 3.0.0","message":"update constraint for os-testr to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Martin Kopec \u003cmkopec@redhat.com\u003e\nmeta: release:Commit: Martin Kopec \u003cmkopec@redhat.com\u003e\nmeta: release:Change-Id: Ia7f0553a3822561b4a3d63643cc0133fb9535dd4\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5baca49822033d63f017d1a200188b4be5e9893a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b0e9db3636448d927bb0c371eee9311715f79db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b0e9db3636448d927bb0c371eee9311715f79db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
