)]}'
{"id":"openstack%2Frequirements~836896","triplet_id":"openstack%2Frequirements~master~Ie32e916fd8298e6ff9631482a73ccf785ec8564d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-04-07 03:23:09.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie32e916fd8298e6ff9631482a73ccf785ec8564d","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-04-07 02:42:36.000000000","updated":"2022-04-07 06:40:31.000000000","submitted":"2022-04-07 06:39:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836896-openstack/requirements/constraints/noclob","meta_rev_id":"f5f9315d226e77e2f3ccb649d5fdbf36d059bfed","_number":836896,"virtual_id_number":836896,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-07 06:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-07 03:23:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-07 03:23:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-07 03:00:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-07 03:20:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-07 05:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54fd8b09a894500a920a3f1b1f1b6858929b5324","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-07 02:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89336e4b93c921f1b7d7470fa91bae21be9a356d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 03:00:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/333dbf0a5b424914bbaff950f8ba569e : FAILURE in 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"cd608957804627df672d164caf713eec37028105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-07 03:20:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65416ce85b51278bfa8775428f9547b28fdfdf64","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-07 03:23:05.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"d86f411122fe9010d5bd6c6d2a45b3bae4eaf50e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-07 03:23:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a860aca825ea8a0e55de50b1e3a695ff3fb8e38d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 03:39:26.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3ab985eab71f4416a252bcb6b6a3dab7 : FAILURE in 16m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8587a30e3a691a9bef584bf9933c1a06a7ec0904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 05:11:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2570488e124744a89123be152da80d9f : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d933de410194bc0b09e15015bfd65c5 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/158b035818b14b2bb029b4b4041ef31c : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74a10f6d7e4e40739494215ee2a1caa6 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea76ae0d6ba14c6d8dba9b767427f2a5 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07cbe377eed54738a426717a450dc26c : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/26a0dff4635d491cbb7c7c0e45a76585 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/54fe0711e2f244d5b15be8b4d4154819 : SUCCESS in 10m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dfd63e3ddd8847849410853178628c9c : SUCCESS in 5m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7397eb21356f4bba9b8a75547ce47f55 : SUCCESS in 10m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/832ebd4b3c3442458ef55d3dbf250c03 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/733e32df11584ca1a99bda57cac4f479 : SUCCESS in 5m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/62ee41053deb412c8e8161a88cae44c5 : SUCCESS in 8m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fafee6b186c44aa2b4501ef81a26db14 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03eddb337a814dd2bdc1b8777ef22ce6 : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/82f6d3e03cca45c6afa0de4a944a516f : SUCCESS in 11m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/89e24565a7a34717a8cc901b106f2b49 : SUCCESS in 12m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b3bdf02d1a5f4be5aba6306d41dc6cd7 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e937878d08f946b78a0c2e11f119dd39 : SUCCESS in 12m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5d6dae41034f4305848360e7c50e0534 : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f7b641ed3d434533a292f2f9389f6bfc : SUCCESS in 24m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/35a946b3e7a1482b98dba6f5597989b2 : SUCCESS in 12m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/30cab7d15d214bf884c906e6d6f8c75f : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6dd1373fe714254a75005754d3406a1 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75f220c617704a99a4503d145ea83e5d : SUCCESS in 18m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d9a73ad90d8c48588d9c19fbf0ae4611 : SUCCESS in 4m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/138b9b0ec85b48e8acb8fc9767edac1c : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69b21ebd5e7c4f2f9b7e46bc4b30c41b : SUCCESS in 6m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b85a583b7426443cb1d4f052ab595346 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a6ad0d792d7433a93e4e37bab6f8fc9 : SUCCESS in 41m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db8b0bacd6fa4e81b7679fa9c6247666 : SUCCESS in 1h 44m 55s","accounts_in_message":[],"_revision_number":2},{"id":"9007f87c72c871b56e203c6e63428298c29cb354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 05:12:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b79dcf54e523267b5073a4c2cafec48865b2f81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 06:39:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a89f8cfeee46cd810d71ced8e4f5b5 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/590f6ea18ff54b2abc03f7426728a06f : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11a4eb33b8f1477692acb7f682cdcdc4 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c07a72cc3464c5ba940494ea51065cb : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d71d498d4df24f0d8201a276e2691d04 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b873dd20fbbb405e873670bd0233026f : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b15e46325eca4d93850992dd0da38af8 : SUCCESS in 8m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cbc399f1d7bd48c2b22808c524d93a80 : SUCCESS in 9m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/26ffbd9beb4149c2b65da781d33af70f : SUCCESS in 10m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/762c923d984b41259ad8bc11a75c8b6e : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ba70b33a69304cbd80e3af461942b816 : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/00fe31c2855740b2864f2a81b25bad77 : SUCCESS in 7m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/596e525dd938484ca4dfcecd65a98130 : SUCCESS in 8m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9fa9505f563c4afb92e292c19e2c4562 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/30823d7a009d4af0a4233af7b801fab6 : SUCCESS in 7m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f8131f1119134ce99601a9e9ecb8aa94 : SUCCESS in 10m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c2d79e5027c04f5b90783c530ab9cb32 : SUCCESS in 4m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/879697743dd74f3ab259ba063fede2ad : SUCCESS in 12m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f6d02663b37b41ee961527b86a8233ae : SUCCESS in 5m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e88437d4ea3b4fb5ae393c3e8684a853 : SUCCESS in 25m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d990b256927c42c8bc48236d0542e326 : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/da8fd90fd09747659f2614ea48b13046 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/772a527447ee460bb23d287d124eed95 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14b854e9245d4ae6b6e3c37e1ca5b122 : SUCCESS in 23m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6028b8a17dde4a379eaab953f4c63d6b : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/abcfe5a0708045ef84334b46ce3959e6 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b6c3e73e5774ae797b4b03b1ef440b1 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f0efff23daec4e458975f8e1f0b27ad4 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3074e2397953440daa88dea0fb1e1239 : SUCCESS in 47m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a78bdec60f3545b09bee83710cdd3f6b : SUCCESS in 1h 26m 28s","accounts_in_message":[],"_revision_number":2},{"id":"cdac429202899668fe27c5b206ca45eca3dfe659","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 06:39:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f5f9315d226e77e2f3ccb649d5fdbf36d059bfed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 06:40:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68d036bf8754074aaaa2d41a1aff829 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b20204939d114c6e9a0574a046bd9a4bb43bbc4c","revisions":{"67f3bfa8170b1bb528d7f7d43f33cb0064e3fbb7":{"kind":"REWORK","_number":1,"created":"2022-04-07 02:42:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/836896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/836896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/836896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/836896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/836896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/836896/1"}}},"commit":{"parents":[{"commit":"f4423fb5e06125e13ba91a0c1f7d3b595517bb59","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4423fb5e06125e13ba91a0c1f7d3b595517bb59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-07 02:42:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-07 02:42:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie32e916fd8298e6ff9631482a73ccf785ec8564d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67f3bfa8170b1bb528d7f7d43f33cb0064e3fbb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67f3bfa8170b1bb528d7f7d43f33cb0064e3fbb7"}]},"branch":"refs/heads/master"},"b20204939d114c6e9a0574a046bd9a4bb43bbc4c":{"kind":"REWORK","_number":2,"created":"2022-04-07 03:20:48.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/96/836896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/836896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/836896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/836896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/836896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/836896/2"}}},"commit":{"parents":[{"commit":"f4423fb5e06125e13ba91a0c1f7d3b595517bb59","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4423fb5e06125e13ba91a0c1f7d3b595517bb59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-07 02:42:34.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-04-07 03:20:41.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie32e916fd8298e6ff9631482a73ccf785ec8564d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b20204939d114c6e9a0574a046bd9a4bb43bbc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b20204939d114c6e9a0574a046bd9a4bb43bbc4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
