)]}'
{"id":"openstack%2Frequirements~837263","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967","project":"openstack/requirements","branch":"stable/train","topic":"validations-libs/1.6.0-stable","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2022-04-27 17:07:09.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-04-27 15:03:08.000000000","reason":"removed on reply"},"23266":{"account":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"last_update":"2022-04-27 17:07:09.000000000","reason":"Change was submitted"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2022-04-27 17:07:09.000000000","reason":"Change was submitted"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-04-26 08:33:11.000000000","reason":"mbu replied on the change"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-04-27 17:07:09.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-04-27 17:07:09.000000000","reason":"Change was submitted"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-04-27 15:05:12.000000000","reason":"Removed by jpodivin by clicking the attention icon"}},"hashtags":[],"change_id":"Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967","subject":"Setting constraint on validations-libs version","status":"MERGED","created":"2022-04-11 08:31:50.000000000","updated":"2022-04-27 17:07:09.000000000","submitted":"2022-04-27 17:07:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"837263-validations-libs/1.6.0-stable","meta_rev_id":"8b5afe79dfa90e6b8ee40e041aaffd1dc2a20f77","_number":837263,"virtual_id_number":837263,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 17:07:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-27 15:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-27 15:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 10:12:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 19:48:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-26 06:29:03.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-04-26 06:29:03.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2022-04-26 06:29:03.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2022-04-26 12:57:20.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-26 12:57:20.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"state":"REVIEWER"},{"updated":"2022-04-26 12:57:20.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"545338253ab8d3465547106e1d171e48ad35614f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-11 08:31:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7a5a8dcf9fe87d86a7e7dedc05e95a4302214b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:12:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89613940153e48b18b2a7a3339c1aa3b : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0dc3efbf87bc4e65ab9d1ec143b7bb41 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de283941bdd24c2fa48ba4f14ea7830c : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02dd5a70f2a4434ba7620f62ba9b1009 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04cd91f13e10495bb408337609f73222 : SUCCESS in 3m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4c227b53b384a7e8154f598b4384a28 : SUCCESS in 1h 04m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb5a4a4f50b848a988afcdcd6aafb3fe : SUCCESS in 1h 34m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/08712d954d874dc0bebb4a8559a6f9bf : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"341c1d93aa2dec4002eb8b35ebeb532a6995995f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-11 12:01:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"377e7d9bf2da648590732d2c7a9e61ce97ede7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 13:21:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1c404c3b414534975e6c73e1dce500 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/630601d673f44a29a846f8073f47b0e0 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a986022dfcad4d86887c71e4720dd06e : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7673be4b691942da82179d4d1bd9df7a : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e05303ecfb794e9da0e1499a521088a6 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a48f819638614450a317927bb64d346c : SUCCESS in 3m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/60072c4104344bcabccacd16cb1dd797 : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de2b958dad674ceeac3aedf773722812 : SUCCESS in 9m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cd37f38594f542eeae45faca8fb8ef89 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6a086db0fd0c444da967eb1be5f91c6a : SUCCESS in 9m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d943856d26f74fdfb62337cd8b9bd7fd : SUCCESS in 6m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a2c29226e97a43aa982283dc0f196c22 : SUCCESS in 5m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ee2ec6b4192d4dde9b0fe4ea3341b3e9 : SUCCESS in 9m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d6490d9dd5984f5ea83bc0659df7ccae : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/516be5cf93f44eedb8771e8bb20365b2 : SUCCESS in 8m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9df4d37e86274169b2d6dee5af46abab : SUCCESS in 24m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64eb89a3f49b4753bb10f46f3481fb7b : SUCCESS in 12m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a58da077831f48f0a1dadca06d872470 : SUCCESS in 22m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7b16ec77a4f545ae9d6b5b9a10203588 : SUCCESS in 20m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/66cdf28ff8d74b02839a1daad1a66d1a : SUCCESS in 14m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4bf60527183b49f2a2df97f19fb0f01a : SUCCESS in 13m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/06bed1053ed9440fa48c911c9f9be65a : SUCCESS in 15m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b859a8a099a4136a8e6801720497cef : SUCCESS in 21m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31ee70eeda5d4cc4abd2a6987fc2e361 : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/adf1939ef044430b9c131a60fd84a5ea : FAILURE in 12m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5fbf1ff2d0684788a46211c5614ac2b4 : FAILURE in 19m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef583184c3e7434ab6612232e8f46563 : FAILURE in 1h 04m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7f6a4f515de441de8072d39222b81a5b : SUCCESS in 1h 13m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/149199779c1346f29db7fd529734266a : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"fc40168db2a7bf97b18a0c2d2c77d1576fb49466","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-12 11:54:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efcde86fa67dbf2c102901732e3a38446b19520f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:44:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b80f2f7b7546a19856ca0134993a0e : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78dec3f4b9774dd9b3b393e5fe26a737 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d027136517ed4ed283bdd637c6cde7cd : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c23542d74ea40e28f1226232ca8b4df : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/477a5917189449999d10912da5a6329e : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5731e6cb04d14b8cb2d8ed823522a068 : SUCCESS in 4m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2e9fd64a58dd449681a8c88b4120ebc1 : SUCCESS in 8m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a053264f3c8648739a0a8841da3d670f : SUCCESS in 9m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/acd6c2feb3834480b297f55907e9c16a : SUCCESS in 9m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7110427cf3c0486c817ba77c9c871db4 : SUCCESS in 8m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/96888fa324df44a1a38fefc42174530d : SUCCESS in 5m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/92c1a464260a4be69458e8a4e51c3a8c : SUCCESS in 5m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cbfc5e3c616f44e093751458938f6ed0 : SUCCESS in 10m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5de3132386a4f5383377d866a7040c6 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18b996ebeb86466193c96034cd7cca89 : SUCCESS in 8m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e32aefeed8434695918dd11863ce1842 : SUCCESS in 12m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/998c1e5ba7924b0891bf7aaf89be5be3 : SUCCESS in 19m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/27c14229b2154de88f2c48f45e690bf8 : SUCCESS in 23m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa267e1a2d5a48f7a17b957e10294d9a : SUCCESS in 19m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8f0875f399344feaab69e5508dce9a28 : SUCCESS in 15m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7e8ce5fdabb047d1b962a318b3a38d57 : SUCCESS in 12m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c0f1cc13adc44853be6226173fc1c51c : SUCCESS in 14m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/677f4cfac9f244d88e1c38b9e8271076 : SUCCESS in 19m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa92d2b0bcd14cdfa4b712ee83b1933b : SUCCESS in 6m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/01689b67b219443480093e2590c054bc : FAILURE in 10m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5368f71c48714fce8527b1e112bee4e2 : FAILURE in 10m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d05dfc59ec4d439ba5ce56b0024e3e5d : SUCCESS in 57m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/414338a88359424398d5b3236dfaf782 : SUCCESS in 1h 35m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/431219f841fa48f0856274c47a505178 : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":2},{"id":"804bb6f8e8e934bb4d3c7306a769d74212193614","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-20 14:47:16.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"72c1810ca34dd15811bdb7425d286f63de3938a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:18:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd934e0f194404cb8be1f9f283e938c : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/056d77af0cfb4020ab79c45a21253d9f : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de22bed021cf42499579255b3260c3ef : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02e7188c7429427c90263387c3aed949 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20212e430b924b5c9b6c6380594b25f7 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f6e94ca09134e8ead594369a446e9ce : SUCCESS in 5m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/151a5979a1c24dfba3f26983f39a66a4 : SUCCESS in 10m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b0745b9aead54c57bfbaab5c997c491e : SUCCESS in 8m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ebdc858976a143f0aed2e51928c062ff : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/049b3d60d5fd421eabaee58b42f9f3fb : SUCCESS in 10m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bdb809209af34eb7bc06982b9aa9d6ae : SUCCESS in 7m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d6dd4e526c34bd98725679cb2fac994 : SUCCESS in 5m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f2ac347dfb8a44d982d2e084f1694925 : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a70eed0409744f5b9a3cabe92b9c462e : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53beac76e55d4401ad90456d39735eba : SUCCESS in 10m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e3b03bf871144cdda6bbc7591fe7c483 : SUCCESS in 14m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8012e84e4be24d82b82b43ab8f9e6197 : SUCCESS in 15m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b9180ed308c54909b43baa882e358751 : SUCCESS in 22m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb933012946b4a1b9acfb889330fffeb : SUCCESS in 19m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f30efd270c634027a4ee9df398774861 : SUCCESS in 14m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9e41fdac4bd47ac9c509eddbd83d07e : SUCCESS in 13m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f8fd85d9e9364a009edf50b1d08292e0 : SUCCESS in 15m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3860e547ab2f4bb4be313ec43a65a6f7 : SUCCESS in 20m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75895596bdb5416dbf61635de82599ba : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/08f02912491843f6ab100832494758a4 : SUCCESS in 11m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b31777120ed0402eb8b1406ea1e2266d : SUCCESS in 10m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccdbbd3e24cb4ef18b887670aa13d5a1 : FAILURE in 23m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dfc7529d621641c1b42b811f0478ce18 : FAILURE in 18m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ddf754af8af468c930850771bf3a208 : FAILURE in 22m 52s","accounts_in_message":[],"_revision_number":3},{"id":"841f0648b01c4f70f18c2cbbbe02c084524e4e8d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-21 06:02:28.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6779516b986fcb7709a89d946aa47814f7da711e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 07:44:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a08a1683d804269be99e0dc7e8d860e : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b724cc2555f49d78b4a8e1a8475f14b : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cbb36d28ce84a8fb71f221708508b63 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6320b33f307e417ea5767fd09b495390 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe6db4b0f0e34ee8bc2042b2a94b44c4 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18e2c4ab1fbe412a974122f44a485be1 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4adbde548b6940dc852080238cf1430e : SUCCESS in 6m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fc8be4099306456f872eb65e3179bf02 : SUCCESS in 9m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/365b5d8a0136434dae800e6a6fcb7874 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/491c491acdcb456cadb9f7f88deb7849 : SUCCESS in 9m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3a950d7bc86c48d2afcfb2c242caf627 : SUCCESS in 4m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4a67f17ebf241a0953dac2144804eb9 : SUCCESS in 4m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f0fec65a3acc4b8f93924f695c9022b2 : SUCCESS in 8m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88e2b7795a1b40d08224bbdab606600b : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8ccdf718fe7465987d4d9514faec22c : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f4dfb8e06e354ac4bb092861959f56c5 : SUCCESS in 13m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0cecdfe037e849a6bdaa17aaaa75c091 : SUCCESS in 28m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2e4ba48c105d46b88dda422b344822ac : SUCCESS in 21m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/efcdda4a687641baa5ce0c33a14f31bd : SUCCESS in 19m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5d8644a4df1148c0b73ef3503ad20393 : SUCCESS in 13m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab7e95b696d1400b880bc5382ed8b760 : SUCCESS in 12m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f647987246b4495c8ee5be77c72b524e : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b60894cc5a2415b92342b091100d961 : SUCCESS in 20m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5e22a1b5a0f404aa9df6f23a601ebcf : SUCCESS in 4m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a6f1bfd6eb754ab0af4ee6bacad20837 : SUCCESS in 10m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/50a77d216648410ab6ec8408fcc0a84c : SUCCESS in 11m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/37ae5d95919c4c759ef2986e6a0094b5 : SUCCESS in 58m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/792afa3fdc9a4d6f9d3f93b36306f645 : SUCCESS in 1h 34m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4a5241e287254ab2b68528eb4da038ec : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":4},{"id":"e8b33795e6ea65f0b27264ef86ca7595ec1e67da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-21 19:48:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e42c7d35e3a7c9633f19dcad8c7d6e0fede36e20","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-24 03:20:44.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"198d94885311ad5b6b04cdccca2470a6ef810817","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-26 06:28:10.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ef76d2aa26696429a2b23cfbe7c6493c6b723309","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-26 06:29:03.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"06c501545315aa67ef142a5d28efbacc3ada6a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 07:52:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065ca38509c44df1a3ae3b0115e89a78 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ed0a31a013e414ba832bd3fb0b9706e : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c66a811d40e44a5ab3d691925469c7b : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9a689c883c2435a86a680cf0d0ed10a : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/670b52ec7a5a4c6d87f67a20c4491207 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82a643d37a384b66b9e642a83f7466e8 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/479036e5cc83434a85b7b2a0984134f9 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f091238ace2b48079f77709a8f31486a : SUCCESS in 7m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/63c2ce23a24c43c3aec81692409fd86b : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/320a836beb1f4bfb8b950dd02cb8287a : SUCCESS in 9m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8bef5ca45af2404faed57dcdb2aac6ca : SUCCESS in 4m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7f37d8cb8bbb4ac59aad62a788c7ded6 : SUCCESS in 4m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5b57f485930b4b6ab8d40071336b2959 : SUCCESS in 11m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fcd7e64914f34c8391045f2e3c98f638 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ddedce7b48d43eaa101693d20743e70 : SUCCESS in 7m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d59bfeedb43f41c0a6d3339191bc6abd : SUCCESS in 13m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ee0e21d7e31d472589fd0d997bfa02cc : SUCCESS in 11m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/82155c07c7884f96a7458677557eb6cc : SUCCESS in 18m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4c08af8c9357496bb0b0c66872ea1b94 : SUCCESS in 22m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/22026e7d28a24bb68e36c536053b86eb : SUCCESS in 12m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5c2695bf08b9487bb87187ce4234ac12 : SUCCESS in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ae17aed81942478da08dcb81e760853e : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcc64a58de8b4e6da5be963a23930a7c : SUCCESS in 15m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/380bebc01f89434eadd608261cff5ab5 : SUCCESS in 5m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c7d647bf99d143cf9559f8284d5f3da0 : SUCCESS in 9m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1311b6a9598041688aec838b045ae308 : SUCCESS in 9m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa086a7eb6d043b8929e154a84ecc600 : SUCCESS in 57m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1cab67c9007f4d24858c46bb990d9c31 : SUCCESS in 1h 20m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3984d409364742e38c0bc25418ee63e1 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":5},{"id":"4971a004f0c323b947e34c54b6c1223fff906aac","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-04-26 08:33:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ae35d3d37487f299a673077262403fb920d6d266","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-26 19:21:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9a169d9ea3336b3717cd2a46976e38df7900c7df","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-27 06:35:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0560aa401c38025ad7454494da4dfcbfaedfcb38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-27 07:07:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"abef2192f205ead3319eef15dbfb1e43cd111533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 08:40:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c559bd92224dc2840ff88b3d4f45e7 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cba0c48f0354830857099bc2ffbea34 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f5f9ba86ea64836a3402863afc59da9 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ae38dcb84e34c4c8dec08f6ac603c92 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b45a57a0280740b1a12d8d99a26ba455 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7aa421ef386d46eebf977e04aa1301af : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5c021ed5f9c04bd0b03ce4bad3d1b1c1 : SUCCESS in 7m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5d16670ce25d4a1890a12b09d2a69d88 : SUCCESS in 9m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/30f08116cd4244a388d56f0c219faa21 : SUCCESS in 8m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/98686b51a5b34409955b6f80dc6052c4 : SUCCESS in 9m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aed8404f1c5243338d1455e732400e3f : SUCCESS in 4m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3c1b6c07a89f425993ca4f0c402b529f : SUCCESS in 5m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1446351ba40b4b5183016b44b802f52f : SUCCESS in 8m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/85760127b5794fa4ade6d3c125ee78b1 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b15609f4fd2345319934244e1bf315fd : SUCCESS in 8m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/941a67215b61433ebf7c784b4478d2d7 : SUCCESS in 28m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15b3e1e121194478b4dba5a0abbd60db : SUCCESS in 13m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8921c629ce9e4212a36665d409a72dfb : SUCCESS in 22m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1cef2afa791e471abc595d7d62e7e111 : SUCCESS in 17m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a9862ed681704832b929081f976dfb0d : SUCCESS in 17m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cc5eeb19cf2e4911b4b15d238cc79aac : SUCCESS in 17m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/81dd3e2731d74fccae1b31d503451969 : SUCCESS in 17m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b576f7a8ca11415d8560ac8af5c5d1e7 : SUCCESS in 22m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6f3f3915a9f4a56b4347a940e03a55b : SUCCESS in 5m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/06118346a5f44475b24f2e1cf5994407 : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fc462012dc7a4691870fca67eb4a5a16 : SUCCESS in 10m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77e7ef59186d4e90832af00c0969342c : SUCCESS in 1h 04m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ad5fe08a1f474ee4b5ee1cb9709d9e6d : SUCCESS in 1h 27m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/615c982dd59e4ddea444249245cfeef9 : SUCCESS in 1h 16m 29s","accounts_in_message":[],"_revision_number":6},{"id":"8e2567a9804eed2694464ec933042f5cd8699f7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-27 15:03:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4e491eeba84039ac5e6a9baad9cbff75679373f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:04:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"63268578c7de5ed1658c65bfe41748a223c35f39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:07:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb199cf6e4b847beb48ae638746abb9d : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0325039edf534b59ba1d389085a2df59 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bedc921fe6d94710a0f1b48fa89dfd20 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28add46d7b774ca79d71d586267a4834 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73cbc42656f649aeac90f852cfe9c187 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df3aae586c5944a2a8e362f6f5ca1627 : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bd22cc01e5a34304b679611943a0b4fb : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8a00ac91bfcc4394b1e6dc9e129dbed6 : SUCCESS in 9m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bd87e76c2a6648d58aa3490d56eb4bf3 : SUCCESS in 11m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a47de78c8b01404ca14bc987ac665f2e : SUCCESS in 10m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/60fe837aceb74588967822d81d125a21 : SUCCESS in 5m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e77c33bae52f421f8b8b8c42c51d921d : SUCCESS in 5m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c96890d0f46049648a057ad821fce5c3 : SUCCESS in 11m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5190b58405074130bbead1d67960fd94 : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/272d8c13b55c4a5bbc3acb1e9498aad6 : SUCCESS in 8m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/430b81abed504aabb467f42eeeede6a7 : SUCCESS in 13m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/edc9544a0c964e7cad49e2d421036387 : SUCCESS in 13m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/38ec1ad6cbd34a6689945e5fd5144937 : SUCCESS in 23m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d14f8152e7141fc895f3ab09939494e : SUCCESS in 21m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c4110e7cc9d54c5dbab49064ff8b415f : SUCCESS in 13m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d6a1d87f2c8c44498e9730d2518c98b5 : SUCCESS in 12m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c0bde55d32f8498bb48cc7137e644f96 : SUCCESS in 12m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d6e46cf0f174dc49fdca47d71dda1e8 : SUCCESS in 20m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e46ac7cbcf34412488b4d8f6e35abbcf : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e3fe3edc4d1a4a33a01854c2a73d7ee7 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e390c89eab84a949bea4bb713a3aaed : SUCCESS in 11m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbe55fcd89cb4e3ba724c9256f139205 : SUCCESS in 57m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8500ba5b33954bb3abc6617c01080010 : SUCCESS in 1h 33m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7c21c28bd46548a6b871c5fa2beb1545 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":6},{"id":"8b5afe79dfa90e6b8ee40e041aaffd1dc2a20f77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:07:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0ac18028c1b9872a9c56480c887636847af4f625","revisions":{"43063dfe1893737317cf6399bc8e8eb765f0026c":{"kind":"REWORK","_number":1,"created":"2022-04-11 08:31:50.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/63/837263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/837263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/837263/1"}}},"commit":{"parents":[{"commit":"ef77a2ae95debb2be64598c1c624d5d4e08b4cdb","subject":"Set setuptools for different python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef77a2ae95debb2be64598c1c624d5d4e08b4cdb"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 08:31:40.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 08:31:40.000000000","tz":120},"subject":"Setting constraint on validations-libs version in global reqs","message":"Setting constraint on validations-libs version in global reqs\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43063dfe1893737317cf6399bc8e8eb765f0026c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43063dfe1893737317cf6399bc8e8eb765f0026c"}]},"branch":"refs/heads/stable/train"},"ffd2bae8d070e572935f44f658d63ba1f031bca2":{"kind":"REWORK","_number":2,"created":"2022-04-11 12:01:12.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/63/837263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/837263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/837263/2"}}},"commit":{"parents":[{"commit":"ef77a2ae95debb2be64598c1c624d5d4e08b4cdb","subject":"Set setuptools for different python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef77a2ae95debb2be64598c1c624d5d4e08b4cdb"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 08:31:40.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 11:57:52.000000000","tz":120},"subject":"Setting constraint on validations-libs version in global reqs","message":"Setting constraint on validations-libs version in global reqs\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffd2bae8d070e572935f44f658d63ba1f031bca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffd2bae8d070e572935f44f658d63ba1f031bca2"}]},"branch":"refs/heads/stable/train"},"986afd8c9975904103700c262c8dc58176f3319e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-20 14:47:16.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/63/837263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/837263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/837263/3"}}},"commit":{"parents":[{"commit":"ef77a2ae95debb2be64598c1c624d5d4e08b4cdb","subject":"Set setuptools for different python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef77a2ae95debb2be64598c1c624d5d4e08b4cdb"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 08:31:40.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-20 14:47:15.000000000","tz":0},"subject":"Setting constraint on validations-libs version in global reqs","message":"Setting constraint on validations-libs version in global reqs\n\nDepends-On: https://review.opendev.org/c/openstack/swift/+/838589\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/986afd8c9975904103700c262c8dc58176f3319e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/986afd8c9975904103700c262c8dc58176f3319e"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"2e3c514f70025b54a34446f541533736a5a5d885":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-04-21 06:02:28.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/63/837263/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/837263/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/837263/4"}}},"commit":{"parents":[{"commit":"ef77a2ae95debb2be64598c1c624d5d4e08b4cdb","subject":"Set setuptools for different python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef77a2ae95debb2be64598c1c624d5d4e08b4cdb"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 08:31:40.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-21 06:02:28.000000000","tz":0},"subject":"Setting constraint on validations-libs version in global reqs","message":"Setting constraint on validations-libs version in global reqs\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e3c514f70025b54a34446f541533736a5a5d885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e3c514f70025b54a34446f541533736a5a5d885"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"59097fa9cc6dcb589517307e16aaf21558cca7d0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-04-26 06:28:10.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/63/837263/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/837263/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/837263/5"}}},"commit":{"parents":[{"commit":"ef77a2ae95debb2be64598c1c624d5d4e08b4cdb","subject":"Set setuptools for different python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef77a2ae95debb2be64598c1c624d5d4e08b4cdb"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 08:31:40.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-26 06:28:10.000000000","tz":0},"subject":"Setting constraint on validations-libs version in global reqs","message":"Setting constraint on validations-libs version in global reqs\n\nRelated-Bug: #1967277\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59097fa9cc6dcb589517307e16aaf21558cca7d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59097fa9cc6dcb589517307e16aaf21558cca7d0"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"0ac18028c1b9872a9c56480c887636847af4f625":{"kind":"REWORK","_number":6,"created":"2022-04-27 07:07:12.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/63/837263/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/837263/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/837263/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/837263/6"}}},"commit":{"parents":[{"commit":"ef77a2ae95debb2be64598c1c624d5d4e08b4cdb","subject":"Set setuptools for different python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef77a2ae95debb2be64598c1c624d5d4e08b4cdb"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-11 08:31:40.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-04-27 07:05:53.000000000","tz":120},"subject":"Setting constraint on validations-libs version","message":"Setting constraint on validations-libs version\n\nvalidations-libs-1.6.* will be the last set of versions compatible\nwith the train release. All future releases of the library,\nmeaning \u003e\u003d1.7.0 are not and will not be developed with regard\nto train relase.\n\nanisble-runner is being bumped as a dependency for validations-libs.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ia23ce6f3d256fc13c6dd4951ca6bf541b1478967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ac18028c1b9872a9c56480c887636847af4f625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ac18028c1b9872a9c56480c887636847af4f625"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
