)]}'
{"id":"openstack%2Frequirements~837446","triplet_id":"openstack%2Frequirements~master~I2e4a138784d1c9e303b457cce3d492c1e5f69417","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-04-19 00:48:37.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I2e4a138784d1c9e303b457cce3d492c1e5f69417","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-04-12 02:51:15.000000000","updated":"2022-04-19 02:22:48.000000000","submitted":"2022-04-19 02:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837446-openstack/requirements/constraints/noclob","meta_rev_id":"c460f970d18923dd5fb0f8f018a8c3072113ed26","_number":837446,"virtual_id_number":837446,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 02:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-18 20:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-04-19 00:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-04-19 00:48:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 03:10:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-12 04:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-18 16:45:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-18 20:31:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"23db04f16c8db7864c63dac6e1586d3eefd6f5e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-12 02:51:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da1cc86980d8b864096ef0fa3c20de77ec8bf157","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 03:10:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ec1134d858e14d268c7e435a5ccbc5b1 : FAILURE in 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"77a205ef4cd424977d7d88d24d4d336edcc4d687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 04:36:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2078057394e4fb6b059005b3e2085df : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4cf78b388f449e19312a191686f625c : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dba3beccd0884f9a9d742429633db103 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cee82c10b80440fd8ee9671cfcdf3196 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0ff3a43246a486297dd7546c58befa9 : FAILURE in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9d639085b0f46b1b33023374c448671 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cd6eb59a61984c518f693d6233e2df23 : FAILURE in 5m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b736eb3caf2f43cea9d9113dffa77a08 : FAILURE in 7m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e3f5b2cd597243ab8d547577f73117c6 : SUCCESS in 8m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/af8b2d9bb1134e749809ba86ba1ff09f : SUCCESS in 12m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/779ebef1d5864103882502b742079fda : SUCCESS in 6m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f7c56506df5d4ac48a316fd9873e8be9 : SUCCESS in 8m 06s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f35770e8bd3d4665bdbb87809338c9f1 : SUCCESS in 10m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3a08cb5d667d48f8b8672e7af4eaca85 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46a778be65884cfabe9f0742ec064fca : FAILURE in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a915f6e90d6b4f5aa145606fa9442a5c : SUCCESS in 9m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8316dbfa4d9349d0bb93d9eadc1808ef : SUCCESS in 13m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/af617299114249408a5d88181364cfb8 : SUCCESS in 5m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/292fd09ed91c4074a8fc5cfc5e07867c : SUCCESS in 11m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fafd14b89b6240119eb371bc8b0b1543 : SUCCESS in 7m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ccc824f823454d3ba8ea36233097ce6a : SUCCESS in 27m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/020765d7e87c45b3b1838ff27a338770 : SUCCESS in 14m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2820d9bbd9d546c8858df975df5990b6 : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/096a6b51ae744f81a8ad3393bdf5a477 : SUCCESS in 6m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db334b1d39994e47a428c6e74644714a : SUCCESS in 20m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e6c8efbc8c854e46a5dd74545bd60efd : SUCCESS in 9m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4eff69cfb83f4eb18e9c5c71aa27526c : SUCCESS in 8m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3cbdf61b79c24c8c9eac9bc6178695bc : SUCCESS in 7m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dac29a8c0ae240709c39fc8668c258c5 : TIMED_OUT in 41m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/330e1d50ac694d749285f367b4ef5aeb : SUCCESS in 47m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/126014c0246c4ca18ae5920550222f92 : FAILURE in 1h 38m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2de7877c2e666929e9805041e47a44c83dfe775c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-13 02:45:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2abda958f07cd79b35ebe6822b6a85b9de3bc48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 03:36:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cfbfea9caa3414a9c132c09a7de3da4 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4eaa5bef49f4c7f837708c24d470bf6 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16253a9c44764e529a79d2e2326096a8 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fde6f1dea8ba4b94b3f50294a3df2a85 : SUCCESS in 7m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b18d97d2ff94856802bc59dc9a6736f : FAILURE in 7m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c6ffda24e4643b49f0e160d1bb08483 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fead086090a54689b92e92bfdfaa9243 : FAILURE in 6m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c1ea888f67a0418397fbc0821a4d62a9 : FAILURE in 6m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/479540c97a564796a7b9c8c914d86a7c : SUCCESS in 7m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/21244a5f63114ee2abd958b4a2cebe07 : SUCCESS in 11m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/492c78f746ce4eb6ab67995f6952d8ef : SUCCESS in 5m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f97b5d3e357d47d1b996b143aaefbfc0 : SUCCESS in 5m 55s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/75beaf4eba854742808f208fdde54345 : SUCCESS in 9m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ed772c9aba7d408593c8338b229d0b14 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97cc9bdb93cb4b7a91ac65c9b0275021 : FAILURE in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ec8592983ad4ce2b335bb41e7ce1539 : SUCCESS in 10m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/643ac79b3f9845758d8e15ebe35d7abb : SUCCESS in 13m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2a556df572354eb0b1e822ba10ce087e : SUCCESS in 6m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6c39ceb7c02b4979ad799e7f1f2ca1c8 : SUCCESS in 10m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1c34abbca06f4c2393559b93f844ba06 : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a3f14519abde492f9d68f19256efa71a : SUCCESS in 25m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/da0620475cfe40e780b1cb31463b4c29 : SUCCESS in 16m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d0115208a9574dd1aeaee01f52165fb8 : SUCCESS in 7m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e1cdd0a7d6c34b219e57046e050cb885 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42242fc499084365b105021f3379f4fa : SUCCESS in 20m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/59a51155c6a74b938a685d55a1e571af : SUCCESS in 7m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/36a3656cb6ea4cf3b0cccf9b9a98e98c : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc6327c6c7b946baa1013a3bfffc4605 : SUCCESS in 8m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b0185c1798474f2c99712140a8ba67cf : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a928135b740240468ee55cab16797492 : POST_FAILURE in 11m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be922908bb73460c958f5d2fd55b7490 : FAILURE in 10m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f6b13e188670a2cd4b61eac7c97f05aa91d7628c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 03:53:00.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6eae8092ce79472a9fd87af5d873d6bd : FAILURE in 9m 12s","accounts_in_message":[],"_revision_number":2},{"id":"32bd4c2c0a11dd066f63ea02e864074621701c5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-14 02:54:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07d641764b707c9aa21e85e856857d15ed9a6732","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 03:37:35.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/173631b7dffa43e3a521873b41f30d39 : FAILURE in 10m 09s","accounts_in_message":[],"_revision_number":3},{"id":"edff5f626ceee215ff0e3a3f619ac6895e6bd9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 05:02:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09cdd44e55d448dc86c8f2d6207cd175 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817c85552fcf4210afec448b088660d3 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b2ef4f57574620874b9b7336210a0a : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2353cb3f1224d1fa0a7053e48c0abc0 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/562f144f4d904eb3a326048dfa960815 : FAILURE in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/469ebea0082a44dbb0d5a6e45f4e04b3 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b29f16959618476086542505a97bb6d2 : FAILURE in 4m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7491f251bcff40dca477e43a42cfa46d : FAILURE in 4m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/de34801d27a44009ad04e64c406723eb : SUCCESS in 6m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2ce7cabfdabf4a3dbb68cede7fd4c5a8 : SUCCESS in 12m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/dec1493a5f1c4c5dbefc9516d6760694 : SUCCESS in 6m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/54df810ad89e40d9a897ba6b96d7efe4 : SUCCESS in 7m 15s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/abdda1bd5a95465484d81692f45c20f4 : SUCCESS in 9m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/85f717c3cf504139b5127b53612161be : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a22b6011996144b38ac9140b5400468a : FAILURE in 10m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/86bddda2c1d04121939dddb3fe80186e : SUCCESS in 12m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/453b3bd8ad544c1d8cc22bf526af6d7e : SUCCESS in 13m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/45a039874d86439b8f5b0a841c2b2db8 : SUCCESS in 5m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2243af40201b48c9b7cf51e1e1fa5c9a : SUCCESS in 9m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3a7d97e92cfa483bb55e6ec07106bd8e : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cb5b972386554711a85d46a40cf6c6bd : SUCCESS in 22m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/798b1b4b1f2b49518179d9941a3b06e5 : SUCCESS in 14m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/15265795aa2b4883bf3b365148ebd652 : SUCCESS in 9m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/94ee8fd4c74a48e7aac9dae6c1495856 : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/494f188ec8364447b1abde2c9a1565fd : SUCCESS in 22m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1ac9295e94864edca210b77078873f53 : SUCCESS in 7m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eef7170d99654acc92585137f1789975 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/34993332085f49829d556870629246dc : SUCCESS in 10m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a8b5f407f94544dbb46c10b58f06e0d6 : TIMED_OUT in 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ab02c7372be42cfb75f4ab9a8ddc6ea : SUCCESS in 58m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/231c119d586a49719762b424c658e7a2 : FAILURE in 1h 45m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a222f6510adbb61e22469b641f480af0bd582de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-15 02:49:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d4fa1d3130d06851c1d1e1046eac96e9b54c052","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 03:05:03.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c6e5c0a831a45a4a1c1441b7877c6d7 : FAILURE in 11m 58s","accounts_in_message":[],"_revision_number":4},{"id":"d7b573229e6994493b7a6f98a0deee5ee882a640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 05:01:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a233b61f4d4224be76ef358ddbf28d : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e99cb2f8f69f445cb004581ac572d4a2 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/181cb12113d14e238101969555123912 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92492c8a2e7a4bcc9fd7499e1d14d31d : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/154e494f500e47958100ff88f0b4e81d : FAILURE in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c64b628046a43b2865e3725f9408ba9 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ddbf82f994840c0a06807090138a235 : FAILURE in 4m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5f07ed222f164d819dc421c9b2dedfd5 : FAILURE in 5m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5b166d04a02e4f84b6bc135b6f395dec : SUCCESS in 7m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0fa338b1afb849bb96ce2890667038ee : SUCCESS in 11m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9128f1400693481990d5bf8686c6ae20 : SUCCESS in 6m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/81d286b9f08448fb84c9cabd45b3c160 : SUCCESS in 5m 30s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ff5146a738fe4872a6fbaa0422e41d8a : SUCCESS in 9m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dd9e9b02817744e3a06c6c899a0a0555 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47f672cf9fb7454ab8b122e2799a2e48 : FAILURE in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8441e389dd024fadbe8e7aff1193b8d1 : SUCCESS in 10m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e3bf30646dc346ad8800b499f7ccd0b0 : SUCCESS in 11m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7a742e25e664443a872af7d89f4abd73 : SUCCESS in 5m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cadc2469d998449ab08d1279f53e2942 : SUCCESS in 9m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/09fa6368e0684288a7906b74f6100d92 : SUCCESS in 6m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a77752762866440f9523fb562436da97 : SUCCESS in 24m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/735d5e9cbc7e416c9e5107b8721e4869 : SUCCESS in 13m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/168b2dd616e3416f8b70556b9f1e6f71 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8cbfb124a71d41e584e51a100a94f552 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7490782d5a8a4ca09d13a2161ca392cb : SUCCESS in 19m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b96e3ff6fbdb47288dbda7ccb6d77ef4 : SUCCESS in 5m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7d1d3db2066d47a4b583ebda4aec194e : SUCCESS in 7m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ae5efae690c4baba24a648f601a5f58 : SUCCESS in 7m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ae19ffdb2e5640ad9475b4427bb08814 : TIMED_OUT in 41m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83dc77ed05014fb8b262f05f32864d29 : SUCCESS in 4m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0df8e9a87724c11a10d8c005f247094 : SUCCESS in 48m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/341fd633f049463396265ec8097f5f1a : FAILURE in 1h 45m 39s","accounts_in_message":[],"_revision_number":4},{"id":"45f5b9545c520f6a7880189873f80123ec22cbc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-16 02:42:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"22bca9ffbf032eec9130c2c039c73f473e783870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 03:02:14.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/76d49a73a4af4cf3bf4e9917a85e8327 : FAILURE in 13m 30s","accounts_in_message":[],"_revision_number":5},{"id":"cba732dbb92d6bcbd72c7e0921bd71b814bb1b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-16 04:21:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac5307e37364ab0b881757ebfcea26f : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c148984ce2149d4b2627354dec7f8a0 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d25d6f17cc38427a923949ecc6de0cec : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/864e10fddadd43adb08393ecbbfb0672 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c05305920b1446cd98ef45a824685238 : FAILURE in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d73797c481cb435b87d9c73233c74252 : SUCCESS in 5m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1b447a430fe41bb8fd10ed0400d1d5a : FAILURE in 7m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/89734db7926943bcbe05c9d16f4bb1b7 : FAILURE in 7m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ba7b6cecdf64c81afda329db332746c : SUCCESS in 8m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/63143701f8604e1383ccd402e7d3e5ba : SUCCESS in 12m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a6824e6dc83945cea210bf307eeb4d26 : SUCCESS in 7m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aa4470b004e440588ea592b7ed974c39 : SUCCESS in 7m 44s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f01d1bb6f11b404396136e085ea42d2d : SUCCESS in 10m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e91c3e9e3258402493493f769be3f38c : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5252fd4b527407792a5decd4e5bb206 : FAILURE in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4861a73f1654de9990b94bc05a0c121 : SUCCESS in 9m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bf5bbc4f33b24e0e94b3cd620e7b2bd0 : SUCCESS in 12m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/147be04f03d8421e81519238dcad2cd8 : SUCCESS in 6m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/89fab9d0461645c19b1bcc2c0c3ce5c2 : SUCCESS in 10m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f5c5c42662f4f35965fe408ccf9a332 : SUCCESS in 7m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/afd3c4f3bcbc4957b492c45ce4ded266 : SUCCESS in 30m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4c33b81d70eb404eb6a70c560a97f4b8 : SUCCESS in 14m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f184135e97d14ae6a532685324c94b17 : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd2a39b90be04f68b8f44289a85bb65e : SUCCESS in 7m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49dc88ab4c1643419105170f6477c4d0 : SUCCESS in 20m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1f049f67c4e34b579aa280e1d5fb6e88 : SUCCESS in 10m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9577251375864e83891832d793e95ed2 : SUCCESS in 8m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93d44f28d6764d16b0eb28667891648e : SUCCESS in 8m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b2ed7f2af87f47dfafdc8c7e8a39a190 : SUCCESS in 15m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e72dcb2d21fc4b0dbb2b5b4b6f5e4915 : SUCCESS in 7m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5713e2f7e5e4eae8286ee4f7599d5e5 : SUCCESS in 49m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70e4dabe160a44f5b84b971c49e5b45a : FAILURE in 1h 30m 34s","accounts_in_message":[],"_revision_number":5},{"id":"791b0437dceb01fcd3d4b6e3586cf322e740326b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-17 02:44:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0b4b716cde9cbe62254e436c1b054a333bcb0d64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 03:01:28.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/73191e131174452b85222e99deb098f4 : FAILURE in 14m 14s","accounts_in_message":[],"_revision_number":6},{"id":"7a532a8039f9ee0fe178b972158eaae3c4240ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-17 04:08:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f35637351d44849443b31c46d669f9 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cb96f873ab497da2b458187e9e1986 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89feb1d5748e48e1920794c4c476a771 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e19506e3cb14b248560ecda427e72e7 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c3e765e1255491e825fbd8f7a2299e1 : FAILURE in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5abb7f01f62c4e369ab720cd262e65b8 : SUCCESS in 4m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8787e77a3844246b4285a3b9bc09efc : FAILURE in 4m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/42939d4deb6e44bcba0ce6e60a9eeb34 : FAILURE in 4m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5387e8d119de4c19927858bcd4bbe2cf : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fa11cfafddcd4f9ea14c9ea7df1169e3 : SUCCESS in 11m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8c645705206944b287c5622fc5b34b08 : SUCCESS in 5m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/89f2a797518f488a883909087b60e4ce : SUCCESS in 6m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6ab11489118c4d139eff9641070aedba : SUCCESS in 8m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2e489d9cd82f40ed90cda9b643b4433e : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4030ed5bad0744e197be8a86cde6ab9a : FAILURE in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/094791d5a0a945ccb42a0f219ce78445 : SUCCESS in 8m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/29d508409b504202af5dda93d44b837c : SUCCESS in 12m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a2c1b17eb8154d5fbf0a9b271181ffdf : SUCCESS in 5m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/315a57bce229421c93a73d23c2a08b15 : SUCCESS in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c14fc50e9e77434eb0fe14b601e52ef6 : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8bf76970348a480a93571423561d1d70 : SUCCESS in 26m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cc4d78b307074f16b37528d47cd7a310 : SUCCESS in 12m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d606fcfc02b34ed9a0a4c44932f90e4e : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bdd4d4fa365b4ef9a88a214043b53d2c : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07ba977c02254883b5b095babe3ad88a : SUCCESS in 20m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/eeaa01e52b8f4f50a4934b18964ebc57 : SUCCESS in 6m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1b36ead868254c7fa85fa3a2a60d8246 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ba6f25f01c04fda98860b8262c0441f : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fde5e13a2b87411783eb8382f3f00a94 : SUCCESS in 13m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb266cf1a234c9e948d7e828c87746b : SUCCESS in 4m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9611b5eccf884ab99eeca6bf7e700442 : SUCCESS in 41m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b334f82c6e64ba6894bda5847dbd655 : FAILURE in 1h 20m 03s","accounts_in_message":[],"_revision_number":6},{"id":"b3a8d2ec26efd0705c2260d118b878478cd153d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-18 02:47:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1a5dd67fb14178c7e8be5739a917d3ecc541e800","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 03:04:16.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1991bd00901041748aa2a60b1bca5b5d : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":7},{"id":"eb4f7f267d16ee99ca657c19ab4b085686ebc029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 04:11:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7308aabef04bee95358ad4f06e3a95 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2aa851eef643459da6ac5f0223847a : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6666dbe693f1419fabb0e1115b631cbb : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccbbbb6265924bcc8e088f62702f0cec : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54483eabd13b45cbbd1ff4bebbb14fbd : FAILURE in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4c804989da04a78916ad32e3ca3872c : SUCCESS in 4m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eac2ec77042b4a7db60a95f73f4870a1 : FAILURE in 5m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/103573cdd0ec4107a413caadf9f173fc : FAILURE in 6m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/600b0e15f44647e1ad51d52c1efef4bf : SUCCESS in 8m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c22074e72b6741cc8ec30b1ae31eaee7 : SUCCESS in 15m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/541c48ffabfd4e2a944994dfdcd97ff8 : SUCCESS in 7m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b0da1f92db084f8bb6aa36416c7a1558 : SUCCESS in 8m 41s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7bd4c32cbb314e0ba732c77b92abe98d : SUCCESS in 10m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d56f2050fd9b4a3395f1cc7671314227 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c443176fe4944f18361e3a738f92537 : FAILURE in 9m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b0a2baace1a14bfdb96e1aee49f8ab1f : SUCCESS in 11m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6ab3ca4e4f2f48908041b41f5a99c4bb : SUCCESS in 13m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8b2243c1e9e648f0a98026d5da5461c7 : SUCCESS in 6m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/98d4798a55164c80abbe9957b02717f9 : SUCCESS in 10m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/737b6622d17c453080d90b50a9fc5c7a : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f5e35c28402f435fa75b06972b904e5f : SUCCESS in 30m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a8d2a34e983a4d01ba9bd8653d4ff3b2 : SUCCESS in 13m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4a23d02ddc8a497886e14a2f5da7f649 : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/529d8cf25cc14de69d6b24f30895ca31 : SUCCESS in 6m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84f767969a024810acbe5d823140ee02 : SUCCESS in 19m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/34b3e49ef6b641609074274e40636971 : SUCCESS in 6m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f8a380decfba488893c850090e2d3f9c : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b650e3031d544b5785b97e20a24dc4e4 : SUCCESS in 8m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5e930a3177564e35bdbed3c58f114c90 : SUCCESS in 14m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9297e18464f64013a07eac97ea8beb02 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1b17122c00e44a681180d45950f4b8f : SUCCESS in 50m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f5bdf11696349b28db3cb920792f76d : FAILURE in 1h 18m 47s","accounts_in_message":[],"_revision_number":7},{"id":"8cbd3806e1d9e11b5f747c6da46ab5479d1c56ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-18 16:45:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"34d3e104a9af0558a950c9e9cb317824e2f85caa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-18 16:46:39.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":8},{"id":"bbadaadc419452b64aededd280025e758c01d9b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 17:07:48.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ae326ad6470040d3a65d2b5480c50393 : FAILURE in 20m 10s","accounts_in_message":[],"_revision_number":8},{"id":"f3ea29bbfaac6a2905c7d7bcc6ee736c019c7e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 18:19:42.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3185989eff24088a4bb29b57e4bc05d : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf7a198cfcd4e1bb3a4d60e36bc5fb4 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e12866018d24435ea9b85f05eb0c44f0 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eb0e8221e68495bac9b334e71d2f44b : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6edb3ba50dc5441c980205d1648fc3a2 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/748a56f18e944a39ae893d27467294e9 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3bf933ba3cb44f17bcaceab3cc4318d8 : SUCCESS in 11m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8f0f6688930747ed9446d3c7f83c1873 : SUCCESS in 24m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f2dcb474a76d4b17bba66024482c442f : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9e4d9d612d4145148bcd3361f05abaa2 : SUCCESS in 11m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8f2cdfdb7a2847e1ba256034d11b1181 : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4026a273b102410997c602e1820abe9e : SUCCESS in 5m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/012ad8f83eff4a6290175e5b837c7650 : SUCCESS in 8m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d364fbae80a34cfa845728f0cd0b54d1 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a41c68f6c0b4482f897db9bb989b2175 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6dddc0525e294db68a0d5aed0e681b13 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e30fa96c0c384f15a16f9c4a78ae8e85 : SUCCESS in 14m 33s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/079c2c476bb349808a0407988b012323 : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d5266078260a415788820add2cf7a0c6 : SUCCESS in 9m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c83ed93b6823405681b333d1c22dc3d6 : SUCCESS in 4m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/da9dedaed0224e04ae1c1fe62ffefd1b : SUCCESS in 27m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8cc8d3e892f0456d8e39cccaeed0a08a : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4fbd57f647894419aecad72a1e3ac4bd : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b8afb85571a466bb3363e0ab6866ec4 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0afa3d8665d4910a6401f5877e21563 : SUCCESS in 21m 42s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4f958a9a8bde4645be599706dfdbbfbf : SUCCESS in 5m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ca70b5ef08df450282d7c73b5107c21b : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c74ee1a379d46ec949d295831c61cd0 : SUCCESS in 8m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e11f8dd608d4b07b510c861cef5f80b : TIMED_OUT in 41m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20855a34c6284de38901cb4ffdc6c38f : SUCCESS in 4m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe83423686334ccf92831440e6ae07f6 : SUCCESS in 43m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44546d3835f446628b1f62f6693d643c : FAILURE in 1h 31m 55s","accounts_in_message":[],"_revision_number":8},{"id":"a6d9a37264ed7463955eee4118b8a299368f4248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-18 18:25:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"34ad8b614f956637de785117bea1b9fb89982345","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 18:44:28.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b0ac3715a64b4b3f87f9462a9df2d633 : FAILURE in 16m 57s","accounts_in_message":[],"_revision_number":9},{"id":"05347e87f8374a4741cc1bc22aa41da8d716c159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 19:34:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56dd082e81664974ba73ca24f2b59132 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d25c44c62a04622bd6a0c4ec8ecdde5 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781183a5d2914e41bb4f8d3c9f908925 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0546c082940c48778b62029e2d17cdb9 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc87f29d54124aa7bab082ac47d841fb : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dde5f1e9b97245e5b1c27a23dff2c0c7 : SUCCESS in 4m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c1ff1cfb3dd94d1ca39e3d7ddeb50f26 : SUCCESS in 10m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fbe7869e9cd94f549c113bdd50dea458 : SUCCESS in 18m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9624b2004db84cfeb3af5d3b266522db : SUCCESS in 5m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/165ceb38173140a3bb2cf790d0552f80 : SUCCESS in 10m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e2f6f6d0001441c0a601ec2c8db96c46 : SUCCESS in 4m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/725f52b173da46bf82cc632f70c1f45c : SUCCESS in 5m 41s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5d7784d6fc15492db0ab7884a5d8f40b : SUCCESS in 6m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dfa2c29e334246f2b5712ac615f4fd5f : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e30c8df3f75474c85a6c4ae8da38e7c : SUCCESS in 7m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d130b3c03dd145ba96138b8599c8316f : SUCCESS in 7m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fef3b1f5884f45159e3ca19a349035b6 : SUCCESS in 10m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f2da11a1c18445b6ae4e60041972f15d : SUCCESS in 4m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5adeb8f9b29246faab95e9ed4004c795 : SUCCESS in 7m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3a25a6121dab4505b20390cd5acbb2b9 : SUCCESS in 5m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2a5deaa1e8dd4439926ace907d971520 : FAILURE in 20m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e9ace296d4db4a4eaebabaa04f7431b2 : SUCCESS in 10m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/59ad69cf4faf42acb14622c3f4f98e72 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dee5631998914994815cc79aaca81c94 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5ed74a1e42e4f3e9424ab395356ea7b : SUCCESS in 18m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b6a8b482ccb64b248d6ccf2504e3f479 : SUCCESS in 5m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/79e3fbe66d464193b1b1fb14bdac60b3 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/16fe7b45daa448189396970c852ded0f : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9bcb3c61115d40cc9b2bb1094b3f0924 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8047f28063543dd8c565802012b672a : SUCCESS in 3m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dbc6f67f08e4787b01f7a741396fb0f : SUCCESS in 35m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b66551e836bf462680cb75e475bfa809 : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":9},{"id":"940690daa4180efd1dcfeba7236439a734edefc7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-18 20:31:34.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3e215986b5a43d969c27d132e621193445525260","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 20:51:19.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0d9b1629c3fd45c382724ae55cfc6e5f : FAILURE in 17m 56s","accounts_in_message":[],"_revision_number":9},{"id":"a33c2d774197c12034ed81ca1dd7cc3bc45c8882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 22:29:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a470d37a644e60a7ae80703386825e : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b07a58970c84738bd885eed40ceabff : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e30f87983c4470ca09c64c4fe4831ae : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/543a420cb8b34375a4028699c30dc64f : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f5f13d498d9466992d4b2a5f245716c : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/69935b754f1a44b8ab978c4705f86d95 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/339a5715804a40f99ad5617cc94d3c3b : SUCCESS in 10m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2ced2e84c554949b2d4308319abf067 : SUCCESS in 18m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e46648d8a12b4d6ca3f1a2a2c0f05a4a : SUCCESS in 5m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/93a8906afc5c42d992e149ce0b2e936b : SUCCESS in 11m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ca2902655e8a4ed79155ef2e4f682369 : SUCCESS in 5m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/19fd76bdb3f0484692e465ac9b4ea350 : SUCCESS in 5m 42s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/03eb69e5d84442d9811307b528348c5b : SUCCESS in 8m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1aa07c022d334de0967a64641d23e50a : SUCCESS in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8122dbb37a34dbaaf577ffdff030b20 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/02d755ed1c5a4a1caf7e0bb3f89ccb5b : SUCCESS in 9m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1450a1cad2434563a937326562e93580 : SUCCESS in 12m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5f2f3040887d43ec8b36c3d5b1e06bcb : SUCCESS in 4m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/de3de8b81394450791f45695cbda4923 : SUCCESS in 10m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f469f49b6892491fb44659c66289b118 : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/88d9e64652ae41458d9d10a486bded27 : SUCCESS in 22m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0f134f54e05f4e6485a93810806bba96 : SUCCESS in 15m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f644bea82cad49d6aafc996b4e6e81ed : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/11be2216fdbe4ba5815b705c28bf0707 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f9d52b853b34ab3a6ce487ea38f1cb6 : SUCCESS in 19m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/24d36296cf734b8f9bb8c1f0a2bc029b : SUCCESS in 6m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b9be0af420894ad0b18de347e9535cec : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b950fe8ef81c4edd82acbf4ba1b35c72 : SUCCESS in 7m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/581dfeee3ddf426bb3abe67d2d49e1ae : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4d48538c4fe4391ab80525e686aeea1 : SUCCESS in 4m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/294b3062ebe64769a7fceba63c382bab : SUCCESS in 57m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0616db349ff4779af63b3b1f3d2c289 : SUCCESS in 1h 51m 44s","accounts_in_message":[],"_revision_number":9},{"id":"616efaa3d824e8a4a4162eb5697bc9d92a76c4e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-19 00:48:37.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"3fb4e771f42fd2b03f040997af0b588c738680cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 00:49:35.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5ff81a4adced41d78e94723de769242584121c06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 02:19:35.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb398dcd7d584115af56f232754e37c0 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb6cd76456b465f9d5e32d40d5f83d7 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f0a3609f64741329060e2c2cf9774dc : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81fb67384f894a189d1ff3947d3a48e8 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e6bc2238d904c3092f049c4f4d9d6a4 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39bad6e858d140b0af5b5adc801a715e : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e7eba6a622eb49d1bbd05bac34be0aee : SUCCESS in 11m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9bfcf45fbc3a42faa56a7b4093c7174b : SUCCESS in 23m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dedf772ca5294bfdb440f239f6f6aa57 : SUCCESS in 11m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6457395100cc4f9992ea136a91c316a5 : SUCCESS in 4m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a40fe4e3f2ab4df2babd47e82ed7b401 : SUCCESS in 5m 07s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a7db6cfe923f4997a0017985dc43a4ca : SUCCESS in 7m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3d980f2dc21f4007becc6b96673d3e92 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d42d15530be43ae81f5bfc2a60e046c : SUCCESS in 7m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c2ff1b9c7d9d4e8db07efc6d022916ff : SUCCESS in 7m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/40b7d2b9cd244f56aa4e8a0962b9b798 : SUCCESS in 10m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1ad9c98a974844bc98f2204f3da0c0a5 : SUCCESS in 4m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/933cdd6edc70472b9d6b45b3cde260b8 : SUCCESS in 8m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2a78a6b0e5984ff985bd5a7de5660e05 : SUCCESS in 4m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/080603c0fbd6429c8cea3cf8ee6ec671 : SUCCESS in 24m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9c009e997af14e088ad08977ed68330d : SUCCESS in 11m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9e68e5d5eec047ff8ac5edd39f6d1fde : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e11dafbe899c484b9f1bb794e4999f4c : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/466e0952e5a34ebe9b9a561bd8d788c7 : SUCCESS in 19m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d03ac34e6fc846de8b7449ead5baec5b : SUCCESS in 5m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ac86ea87fb664f95a4f60e141b05be44 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2bd349add8394162ac428918713bac8d : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/408e47edbf10482cbf20a6790adb887c : SUCCESS in 10m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15d39de69c8a43b5a83533cc6b2e832a : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40dc24f8751a4fcfbe3237d2f5f65fe7 : SUCCESS in 49m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efd62519d3a0402a97163eb4c812a2f2 : SUCCESS in 1h 26m 05s","accounts_in_message":[],"_revision_number":9},{"id":"a6652fa055e12a19a091713429ad154f98ee0cc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 02:19:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c460f970d18923dd5fb0f8f018a8c3072113ed26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 02:22:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b00c690215b44279b15d0d1f6bae61b : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b3bd4904b899846595c8cb076c3630e3203242ac","revisions":{"ba68a32dfb4cd707a720e6bca7a4a2f86f1dc9d3":{"kind":"REWORK","_number":1,"created":"2022-04-12 02:51:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/837446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/1"}}},"commit":{"parents":[{"commit":"7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c","subject":"update constraint for ovsdbapp to new release 1.16.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 02:51:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 02:51:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba68a32dfb4cd707a720e6bca7a4a2f86f1dc9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba68a32dfb4cd707a720e6bca7a4a2f86f1dc9d3"}]},"branch":"refs/heads/master"},"9e9839ef81d0e052dafba16da8a1afa3f615f959":{"kind":"REWORK","_number":2,"created":"2022-04-13 02:45:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/837446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/2"}}},"commit":{"parents":[{"commit":"7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c","subject":"update constraint for ovsdbapp to new release 1.16.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-13 02:45:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-13 02:45:28.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e9839ef81d0e052dafba16da8a1afa3f615f959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e9839ef81d0e052dafba16da8a1afa3f615f959"}]},"branch":"refs/heads/master"},"7028596c85c526976a71d80ad240634204ffc60c":{"kind":"REWORK","_number":3,"created":"2022-04-14 02:54:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/837446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/3"}}},"commit":{"parents":[{"commit":"7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c","subject":"update constraint for ovsdbapp to new release 1.16.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-14 02:54:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-14 02:54:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7028596c85c526976a71d80ad240634204ffc60c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7028596c85c526976a71d80ad240634204ffc60c"}]},"branch":"refs/heads/master"},"ca31dd1fabef64c34a12fe32798006203c378512":{"kind":"REWORK","_number":4,"created":"2022-04-15 02:49:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/837446/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/4"}}},"commit":{"parents":[{"commit":"4f4dd0961db0db07295ea3938a4cdcbfad4674e0","subject":"Merge \"update constraint for os-net-config to new release 15.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f4dd0961db0db07295ea3938a4cdcbfad4674e0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-15 02:49:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-15 02:49:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca31dd1fabef64c34a12fe32798006203c378512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca31dd1fabef64c34a12fe32798006203c378512"}]},"branch":"refs/heads/master"},"bc5505a3758390142049cf2c56a3c0258046cf9e":{"kind":"REWORK","_number":5,"created":"2022-04-16 02:42:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/837446/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/5"}}},"commit":{"parents":[{"commit":"53a2f8d4dd986cb5547b5ba48fe3765e654c246d","subject":"Bump diskimage-builder version due CVE-2022-24765","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53a2f8d4dd986cb5547b5ba48fe3765e654c246d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-16 02:42:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-16 02:42:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc5505a3758390142049cf2c56a3c0258046cf9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc5505a3758390142049cf2c56a3c0258046cf9e"}]},"branch":"refs/heads/master"},"86b6fa6998e93b47cdfb80d42778912fc099eaef":{"kind":"REWORK","_number":6,"created":"2022-04-17 02:44:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/837446/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/6"}}},"commit":{"parents":[{"commit":"53a2f8d4dd986cb5547b5ba48fe3765e654c246d","subject":"Bump diskimage-builder version due CVE-2022-24765","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53a2f8d4dd986cb5547b5ba48fe3765e654c246d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-17 02:44:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-17 02:44:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86b6fa6998e93b47cdfb80d42778912fc099eaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86b6fa6998e93b47cdfb80d42778912fc099eaef"}]},"branch":"refs/heads/master"},"67fcf69a2016d5cce2abccbb371df65ce77d2ad6":{"kind":"REWORK","_number":7,"created":"2022-04-18 02:47:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/837446/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/7"}}},"commit":{"parents":[{"commit":"53a2f8d4dd986cb5547b5ba48fe3765e654c246d","subject":"Bump diskimage-builder version due CVE-2022-24765","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53a2f8d4dd986cb5547b5ba48fe3765e654c246d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-18 02:47:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-18 02:47:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67fcf69a2016d5cce2abccbb371df65ce77d2ad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67fcf69a2016d5cce2abccbb371df65ce77d2ad6"}]},"branch":"refs/heads/master"},"fd589c18cdd58021b604fae6b9a52d66883830ed":{"kind":"REWORK","_number":8,"created":"2022-04-18 16:45:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/46/837446/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/8"}}},"commit":{"parents":[{"commit":"53a2f8d4dd986cb5547b5ba48fe3765e654c246d","subject":"Bump diskimage-builder version due CVE-2022-24765","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53a2f8d4dd986cb5547b5ba48fe3765e654c246d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-18 02:47:44.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-04-18 16:44:50.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd589c18cdd58021b604fae6b9a52d66883830ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd589c18cdd58021b604fae6b9a52d66883830ed"}]},"branch":"refs/heads/master"},"b3bd4904b899846595c8cb076c3630e3203242ac":{"kind":"REWORK","_number":9,"created":"2022-04-18 18:25:21.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/46/837446/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/837446/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/837446/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/837446/9"}}},"commit":{"parents":[{"commit":"53a2f8d4dd986cb5547b5ba48fe3765e654c246d","subject":"Bump diskimage-builder version due CVE-2022-24765","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53a2f8d4dd986cb5547b5ba48fe3765e654c246d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-18 02:47:44.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-04-18 18:25:10.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2e4a138784d1c9e303b457cce3d492c1e5f69417\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3bd4904b899846595c8cb076c3630e3203242ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3bd4904b899846595c8cb076c3630e3203242ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
