)]}'
{"id":"openstack%2Frequirements~837596","triplet_id":"openstack%2Frequirements~master~Ia3988bffe838ecc9db1bf781c432166e113f8d57","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia3988bffe838ecc9db1bf781c432166e113f8d57","subject":"update constraint for os-refresh-config to new release 13.1.0","status":"MERGED","created":"2022-04-12 16:36:36.000000000","updated":"2022-04-14 19:09:48.000000000","submitted":"2022-04-14 19:06:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837596-new-release","meta_rev_id":"ec7bea65822c49c4bccc2814b98062ee0bc14e8d","_number":837596,"virtual_id_number":837596,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 19:06:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-12 18:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-04-13 00:20:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-04-13 06:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-04-13 00:20:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 16:57:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-12 18:13:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-12 18:27:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-13 00:20:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-13 06:50:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2ca03c76e47953def8b2ead5eb439cd3fe311873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-12 16:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70fe4ad578ea82beebb7d231da234b0a8b1e836f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:57:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/674759cff3d14af8bec1408ce8925f7b : FAILURE in 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fb34f25ac56082c2ab51108db16866a63704c898","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-12 18:13:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a53b1418f1b4b7e852b6df29720dc33eaee2e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 18:27:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883d0da7736749cc972160ea84dc39c9 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ca7f5896cd49ee9edca976a9a4b97a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/224be4448a474eacae34057a55ce8d32 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/069b8be787ad4d28a55916420c0a711a : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e16a9f9f30554c18bfeee39aae0b26ae : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd2202e43d6247b0aafcafd4c87512e7 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e20addf1924f4bb1ada5e235ac08d602 : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a35001caa5ff4ef48e8564c4f7a173e4 : SUCCESS in 14m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a599ca455cef456da10e93cd7a4059dd : SUCCESS in 6m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f0f633958363433ea2bf85006fe44e50 : SUCCESS in 10m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4cb58a93d16e401f8f1f159e47bcd252 : SUCCESS in 5m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d242f5a51a584dc8a0988cb94aee44c0 : SUCCESS in 5m 12s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3490693cf5c34623a04073f93087e6a2 : SUCCESS in 8m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e44e750a570e49c081b43d6a0a80d39a : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e73950f1512842b68adef49c5038a59c : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/87becf1c4972409ba282cbf9bfe9c07f : SUCCESS in 9m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/20af828161264abbb643c69cf21cc6ea : SUCCESS in 11m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8218b1f76cb34ca2be47c87290c6d283 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/01c6ffeeda034478835f40fb245c0791 : SUCCESS in 11m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e278742e80a245a59b84629f544574be : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5fc8f440a7ec453aadea3c6f1a9a6933 : SUCCESS in 24m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4abb999449e2490489a082cb6cf69920 : SUCCESS in 13m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a89cfb6c21574894873aa7314ad85175 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31cafd7d8f9747c6a2505ee7c1dd08d6 : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee88395df199499c8e30d7c2af0b42f5 : SUCCESS in 22m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7a78d6fbe90a401da2837a50ea6e7756 : SUCCESS in 5m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5692ff42177840179360db82de184d06 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/51c2a064b97b431fb644119339a51a60 : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/20f1a91b147746469950c1ec0f331d4b : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ad6c116039f494d86553e2b710351e1 : SUCCESS in 43m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9fdd9d228054afdb2ff80616f08831d : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e722b2b33e43e487d63352ce69218524a61b1548","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 00:20:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c061c65c565b52db272d83b113755054a63f2066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 00:21:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a54454c72aad628560ce2ae3a99fc27f40226b34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 00:56:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69db9aa4637449fb751d6e57ef93cc2 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c125bba407d441108ca89f89d5b0af67 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f66705e3b7da466eacda1debd837d32c : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63f91a0208914ef6ae1e627158623060 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d1584b562564f229da8d16e1681335d : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/080d61557c4e4579af30ac7f085b21c2 : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1c4e6f5713554bb69b3719e8399adb6a : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/baba3643f53747bbab3a0840cf0680d2 : SUCCESS in 11m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d3a0531c9c0d455a9181064b1430824d : SUCCESS in 10m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8e6a5fe8a7ac4a03a3f70e2587679c05 : SUCCESS in 5m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/366deb56be5d4fa58f36469764edcb06 : SUCCESS in 5m 43s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4040e43128c546bdb3a6714c927da7ad : SUCCESS in 8m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/79e1c3e2fed440338317f69990f527d0 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/149e7448308b4446a0c1273ac26dfbb7 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2989fd3f939143829664581a060b1e30 : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/47960af2dac64854b2b02150a01697c6 : SUCCESS in 11m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bbca0fe57ab24a5ca4cd87f61cfe9064 : SUCCESS in 4m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c7c9253445ea42bb8612e4c2cc6724e2 : SUCCESS in 10m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9e87bff949ba4ea4a59a53f55ff4f9d4 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3e872ed6ccf046bd94040fa106366624 : SUCCESS in 27m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/563f0aad5d9a41b3b45bc507ec200e07 : SUCCESS in 13m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0ce704382ca0439889b566f906753a69 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f79839ce8e56472193f007b1bb4d8725 : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b391d8dffc24612ac102572e7bce8a4 : SUCCESS in 17m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9144b28474714526a1dbb67936ef95f5 : SUCCESS in 5m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9a75c4f740034ad3833c24164ff1f066 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c96a97c227e4dfaad98aa8f5c8a149e : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9b657d076faa41579352af11320f6519 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/805695151841463da8950f993e5aa3e8 : POST_FAILURE in 10m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f2fb7b1c81c4234a583c2689e92bc47 : FAILURE in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"62b4772ceed8a515cd32687f3cd3b055d8d5790b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 02:01:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba03aeb0e04d328b999a6d4342c9555dab73b652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 02:42:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2ddb72d2cd124ee19c03d3c174e605e6 : FAILURE in 19m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f319865a67629c372cf57c531c775a93a19f623d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 02:56:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd74fcd5a6834dbb80e8205e04e1dd5b : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc7901828d444e3885b8cccf893b1f94 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b67d77d0abf94c5781a16cda0ab449c0 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05e14fff546f4e36ae9fb177b507fc67 : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28ba735231db45f699a7a53383a83185 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad020f270c234e6a8ec45022f433a014 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a03c41f600a44d859d5f5c7da5333c0b : SUCCESS in 10m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b2e3433c60c1465a926748fa1d89d2e2 : SUCCESS in 11m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f1d93be0265640598425f9e42fc0c325 : SUCCESS in 8m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/95f31b092ffa4855bbdd4da652be4d06 : SUCCESS in 12m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/79d5cd3c33a34dd1973f3caf544c4358 : SUCCESS in 6m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6ae27bddc9cf4c6c8817bb05935a3ca2 : SUCCESS in 7m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7672c165b36e4466a37c856fd1a1ed2b : SUCCESS in 11m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/97cc657c8e3b4d8c88d214ad8328f76d : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7139516b18dc457aafc0a787b2892a6b : SUCCESS in 12m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c0d794ca6d034ec69d5caa3dfc278a0d : SUCCESS in 11m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/792db54649b041df9733b6973ee3ec13 : SUCCESS in 20m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b3e2d07ed2c74be798561f402dd341ad : SUCCESS in 6m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9c3640ce47134338b430294675aa9934 : SUCCESS in 19m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1720bca95b2441a39fdabf921212e7c5 : SUCCESS in 7m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b47a1ae8a15d413bba1e91e7caec2205 : SUCCESS in 24m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9308405b300d407498218f96ede6ff9c : SUCCESS in 13m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a417406f659b40ca849a288fcc695cbc : SUCCESS in 7m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88281497656e444aa921b62c987a591d : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b71188bb8084b798a4a8af99457fe5e : SUCCESS in 21m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/211b93632ac34d8a99da7aa38e949fa2 : SUCCESS in 8m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/47f20d90ea83437dab26f45b81b8b0d3 : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f4781d9995b498abe21bdd0de376ef6 : SUCCESS in 8m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f4008623189e4e1e9f3095996a0e1b8c : SUCCESS in 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/732e37dcfb6e4b8ba1117216278b0a7e : POST_FAILURE in 13m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62b6e48208ce47b4ad4c3537fc3eba13 : FAILURE in 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"060fae7902216f3a43f045c466122c14b1e37859","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 04:39:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0c6f3d8aceea41e13508aaa48b82867c07893083","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 04:55:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f1f14faa0b70465db992a0dcf72f04a7 : FAILURE in 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9e6e554daded59e44506301b29b04725a3ee8b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 05:17:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7704b018d54eaa803747e9398924b3 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631c40e9d959420aa0c74f13591e0454 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721bdfbbeb7d465bb71d14952c5de3ae : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03d91677a1d3415ca369e4d8490f8df3 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5acec74f38824423a685d35be558754c : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54cd5af68fdb4d3b8e007a43de4c1386 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72bf44f7b1a94d15b695edd7d0b4878b : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0106cef91635457c9f02501f17475252 : SUCCESS in 11m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e9417d2892f6417eb9b42f4369a25113 : SUCCESS in 5m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f8ea9d3735784b6784d4d8c627bb6023 : SUCCESS in 11m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f0226c2a0f1b44188a0299b992556483 : SUCCESS in 5m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/468cab73ee0e4a22a46f2f4671c9b69d : SUCCESS in 5m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5b385c94419c4b76b3c5c5e06f76a39e : SUCCESS in 9m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5be9c34f5d7a4f379297535eff9698b9 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df50851b95864d5c8f5eea2c83eebb15 : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bc7dabed16ad49a39168c6123d1e11e8 : SUCCESS in 9m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/eb6b91765e8947cfadc846b24e884f25 : SUCCESS in 11m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0be9027e700649d2881bb390addf5bd8 : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/430dc73f282e4231bf37e3cebefe109e : SUCCESS in 9m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e2810a3c6f2544db8a3c10cba513764f : SUCCESS in 6m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/65dfa92e99054ab9abe8ad3f207c46fd : SUCCESS in 30m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/25356a912b2e4517a65cd4896b948184 : SUCCESS in 15m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1b154692657d46258e772d0a5f72455c : SUCCESS in 6m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2db68fc469045a68119836c3f055237 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fb6660c4a8d4d168017ed09607d973f : SUCCESS in 18m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4dcab4855db9451bb339d184ed501724 : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/85e0a395d986427688ff3f4d7e4215f8 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df9e2dca624b4151a275f8331c64ca96 : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/efe8debf3a0243e6919f574f8199e7aa : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90ba168e8cf24259b35e3b65d321027a : POST_FAILURE in 18m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b158c146de154738a48014929e017891 : FAILURE in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"69bcc83aca3bfa7735ae4b380012faa8b3ebc7a8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-13 06:50:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84b52a9341199f174800cdbf4dee8f7d99d9d653","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:17:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/247b66aae31a46f2806da1760ae84156 : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3a0d82d43deea3a40889bd5471f22c6aaad4653a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:30:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3065665e92c44c738f18408a88eabdf0 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613f1d527cb8432c859d434bb19ef18a : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa8f1dbf70b34648af5083441233b3d1 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3489ddc60c6d49b6ab337f2497cdcebc : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1e7d2be3132402486a4d1b79fc6f589 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36046229232b4c65a488963eef18da97 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81aef1f45a394e9b850951cc8b34b09a : SUCCESS in 8m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/19db2136b51d4d57a628456e59b091eb : SUCCESS in 12m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ccddfa2c239c41728ad1550a83ec144b : SUCCESS in 6m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7395e18e584e4415900cee4d90c6a8b5 : SUCCESS in 11m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5f65a694a01e46ce84b2117b58076b54 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/116e1cffa9274fe58cf8048d931c41e6 : SUCCESS in 5m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/dd7f737d607d4f6d8c6b2f2ba6162c38 : SUCCESS in 8m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/366bc8e95e6d473c824f51dcdb44adca : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5addf04bed674ef796caf4709e249517 : SUCCESS in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/35920c30fcce48b0b903a4ecab5112c9 : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f37071dd8dc243e3b655e4deea6a08e9 : SUCCESS in 12m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/075536cf12b34b1eb4e6bea4a72215bc : SUCCESS in 6m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fc1875ddd731466394cc992c2ccf265f : SUCCESS in 10m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6e3e462128e7423cb0e369092974b9fd : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1b3fef04b98848219f7814ffa7ae9679 : SUCCESS in 32m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3d52656c3ea3488799b0fa53389607f1 : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2f92148083914a39b7c2ae6ff4598ca7 : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9331d745a3eb41fc87854d9c7a92509d : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/348b77b376284cf291da57bf8ba8b8d1 : SUCCESS in 23m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2252a50f8416437da42e378e31c5de12 : SUCCESS in 7m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b0e0431b73f64e2786754068bbfb81b1 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/634514e5b8c1479297e3a124967ba847 : SUCCESS in 7m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f5cfcd2c09124e0592a432acd9a3476e : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ae69724491d4e9280dd025c217fb34c : POST_FAILURE in 12m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25a27fabaf7443e5bf0c8ba0aaff3a45 : FAILURE in 18m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ad9a4f9cb88f9c4c98a553bea49ae349e7a5a88b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 15:01:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28f99b9c27d04f276d44d3777576dfbe11fc94bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:23:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f0d376b715614547a6a7565c00d4c75d : FAILURE in 20m 23s","accounts_in_message":[],"_revision_number":1},{"id":"40e372de960207d25b5f9adb1a84ee87e533252a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:41:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1d67a72058549faadde6e4bde1f1747 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96f34d84fad8477787288d2c6f1746f7 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fa6b5c6122e4eaeba152c9a186353b7 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/490c6e7385904b179f08d0112c081748 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d567a589d6004066b11b5a13f275bd3f : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3397695b8ad471fa3701d6166f66a6c : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a4371159d8f4315abad4595dc1b0c5e : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64b36d85d05e40aa887b170b422962d8 : SUCCESS in 12m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6665e37bf04b48238add6c98bff2327c : SUCCESS in 6m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f7427a012e5b411c974eee0208afb332 : SUCCESS in 11m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d46575008751480b8157b2de65e75463 : SUCCESS in 5m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/99cfcdf7958942be915234ca7e659c2f : SUCCESS in 5m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c4ecd068a32e426c8eb83e2b9b7e3452 : SUCCESS in 8m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7655fa98ccaa445ab31dbadc9ec3ebed : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04af4d62e020443facb8a2c2456ac438 : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/050459fec94a47ce98e93574e6b4446b : SUCCESS in 9m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2c21ec23660c4fcda06189ecfa1a8aea : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a1a3d8649e8a460fa92d117f0d2e7f9b : SUCCESS in 4m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ec365b27a03c4adf98f02d405545517f : SUCCESS in 9m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/47273311b1c349f79185bf115471108f : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ad6250b2ef6b4e8784478a0e687ae174 : SUCCESS in 23m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dd07e620fb234b4cab4a379484cdd66c : SUCCESS in 11m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b57c2188974c4592b6e7ca339d176c81 : SUCCESS in 6m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21e97b1188a047aa83f0f93e301ed9fa : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d1665e895864185a42737c9d7c34854 : SUCCESS in 18m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ff227e4e497545c2884af22aa44f32b0 : SUCCESS in 7m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/baf176813c50459d91617ae8109065cf : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6bfc8a07616444e93c80c6e5a764f27 : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/86365e824d0642f5bc42468eba49d20e : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c340b8ea4624432dba9a6084ebbc3df6 : POST_FAILURE in 12m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/340b249872fb4d0894a5ad65e48067e5 : FAILURE in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"48434ee53f18d3b0034032433206c0003e5c8908","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-14 16:08:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0e49a97b02e28454acfa1a3b3c0c6288cc5a486","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 16:54:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/39d71717c9d845e699b0ad5964fddbf7 : FAILURE in 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9db8cb65453f69ed1992b11fa3c029996b375708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:36:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4cef8f3ea6b486dbef1dae678f7abf7 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed8cb4ac3f447b4b69f7641dfdf0063 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/899b6077496f4599924586839de37cdd : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b60e8f24c4764e8083fc4971a3aea831 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6964825f7e3f4eb6a34e284a4b64f921 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/698ce0e537044ef9b19d04d845416710 : SUCCESS in 5m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c7460514d3b4b7fab598b31c35cfade : SUCCESS in 9m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/29e5c49251934c52b0cbbd9f4f5f0915 : SUCCESS in 19m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6b22a1c9357d4e9e94b787a6ebbd42b0 : SUCCESS in 6m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cf1dad0a13d24778a7337e8b9cb313d6 : SUCCESS in 10m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b7d599dc2ca242938dff6145880f919f : SUCCESS in 5m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bf5ee080e8e34059b12100bf8861cd2c : SUCCESS in 6m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ecc81853c48e458b8ad71f1102d2f197 : SUCCESS in 8m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d58bd6fb81ba4c1d91f0effda83f82dd : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/712a9e8858c64b51af27e0701f1dd863 : SUCCESS in 10m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cc5c1b5e96d14c458a0e6fcba4161e22 : SUCCESS in 11m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6b468d42f6014aaabbe59c9c28ccae3e : SUCCESS in 13m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cc9f61f2ed454830b05c6babc3fec9f2 : SUCCESS in 5m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/300c9ce17cb740beaa42741e4be9db7f : SUCCESS in 10m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e3921617c74e4e04bf5a5de019af1228 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/edf78f08566249df89d50b0f2997a676 : SUCCESS in 23m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/03dabe4d87af405f8405508b094e7854 : SUCCESS in 15m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ab8df9a3fdc7487eb39e757a5301b5c5 : SUCCESS in 6m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0b2c355d858c4c67897125683c08982c : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35a0f3d411304630ada73f57e0afac4d : SUCCESS in 22m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cfe4c3ad6caa49cb82f4e7da5051e93d : SUCCESS in 7m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/095df9694e3b47a28a8ec7ce22711c66 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c3e4d355c114f3f9b1e8de05b9bf303 : SUCCESS in 8m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8676e0abe02b4a57b3aae5a762a48e2b : SUCCESS in 12m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6963a945ba5140a6afd28753f855a695 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d06e831ae2864739a2288f344467a29d : SUCCESS in 51m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a05d3ae8e4e46efbfdd5a29188e108f : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3c0693304198f28e90ee47fdc94ee98ac052f1ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:37:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa37883a688cbec371ca5ba658b0145e355dbf59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 19:06:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26687199a97437da2a18a759f7df135 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e0088fb5634cceb0addf2710d0e6bf : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc473544197b4dcdaf418f20d38440e1 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be99ddcae1594b0a892a04e743f0fcce : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06cd2c28a2294f679c4fbd87b8cdfe18 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/427b1ae0e01748538d3f502196e909e8 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8e85e71007fc4a3590c108b5679203aa : SUCCESS in 10m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e454f897190d4c40bbdc70bf6a7bee03 : SUCCESS in 22m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/13eae7038738418dbf5f6dc622f007f1 : SUCCESS in 10m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e01edfb98c264dbdaf7bbd15ca1c95c6 : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/75db2224a767487b8538c5c6adcc375b : SUCCESS in 6m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2d644975f9e143c3900852c65599caf5 : SUCCESS in 8m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2baae143f1704c92b26251a45fd909f1 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/047aab35cb764430877e63e13a03d3f2 : SUCCESS in 7m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5ec18da640054f3ab12563bff5b370fe : SUCCESS in 9m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7e259f3502704dc9b60c774c3565bd21 : SUCCESS in 20m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a491a59a45774d988dcbbdd20ad0f400 : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d36434c86c574ebfb5e10d7225de3b0b : SUCCESS in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/825c31e64f784de99b8d8a8b04c660d6 : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cace94a3f9904657b0ee8772c95c1989 : SUCCESS in 27m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bfe06ad16e4742cbb76329f2f5771e33 : SUCCESS in 15m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e0c9ed20f7034a26883280d73cc9ca19 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d3a4b1cb49647f3892fcb5abb631c5f : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbf70aba583042e1917283167b431996 : SUCCESS in 18m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4ee7b22b6f0e404a815a2d2c6e7a8971 : SUCCESS in 6m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/36b8dcb7ce64473d8a2d292b2c285596 : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc8f05d3f6f94d229d0365c07b6499f3 : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e1b4a13c92f64e908d6ca4fd245aea77 : SUCCESS in 17m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c7a51f98fd74050b931507d351b619b : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cce30b3760f941db8f0f512b3006a080 : SUCCESS in 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f280751ceeb9424191f11e2d8932da26 : SUCCESS in 1h 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e69c24a0cbd1034e71dd446de89005003f4aca81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 19:06:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ec7bea65822c49c4bccc2814b98062ee0bc14e8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 19:09:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c195a8485744a1b86cbc2164f3c90ac : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2b7a4c0e3fe46caad1ba2b936acdf26ce784dd3","revisions":{"d2b7a4c0e3fe46caad1ba2b936acdf26ce784dd3":{"kind":"REWORK","_number":1,"created":"2022-04-12 16:36:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/837596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/837596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/837596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/837596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/837596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/837596/1"}}},"commit":{"parents":[{"commit":"7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c","subject":"update constraint for ovsdbapp to new release 1.16.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 16:36:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 16:36:35.000000000","tz":0},"subject":"update constraint for os-refresh-config to new release 13.1.0","message":"update constraint for os-refresh-config to new release 13.1.0\n\nmeta: version: 13.1.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I8f55b4f4011ee351b0f2eedff8e5d484a9592182\nmeta: release:Code-Review+1: James Slagle \u003cjslagle@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia3988bffe838ecc9db1bf781c432166e113f8d57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2b7a4c0e3fe46caad1ba2b936acdf26ce784dd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2b7a4c0e3fe46caad1ba2b936acdf26ce784dd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
