)]}'
{"id":"openstack%2Frequirements~837598","triplet_id":"openstack%2Frequirements~master~I5b7096e38e7c2d6a567528779d71df7fe504809f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5b7096e38e7c2d6a567528779d71df7fe504809f","subject":"update constraint for os-net-config to new release 15.2.0","status":"MERGED","created":"2022-04-12 16:39:34.000000000","updated":"2022-04-15 01:10:45.000000000","submitted":"2022-04-15 01:09:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837598-new-release","meta_rev_id":"9e1aeb0e594cc4eb32f22fa5fdfcb62ab303d786","_number":837598,"virtual_id_number":837598,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-15 01:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-12 18:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-04-13 00:20:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-04-13 06:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-04-13 00:20:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 16:57:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-12 18:13:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-12 19:44:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-13 00:20:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-13 06:50:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"c484c2d851df0579259b25e625253d50d5b140f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-12 16:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3ee4a4c6ffb9a9ad8c0968f64471433f366bc23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 16:57:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c874699500ab4b60adc76257da8ef9e6 : FAILURE in 16m 02s","accounts_in_message":[],"_revision_number":1},{"id":"88f1cd3493f85b6c672bc220028ff0c871e3043f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-12 18:13:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1525d7973a72294df09322ba0f87499449b76a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:44:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bbe2673c8b44450ac72c8fef31706ff : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ad4caab200441ab35962437753bf40 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cc9ec379fae44be895b42e615f0ee54 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b10fbd1c85ca4010982d3c2c50ae0392 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91da422eb1d74419bae305f8e15e8b68 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/706cdc98bed242699415e7be742c5d16 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/234cc477b3ac46bfa136a0c32b38204a : SUCCESS in 10m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3aee7525b3e745b081bf2c5954a7ed77 : SUCCESS in 11m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2054630fcf8b409c98d67c7e9c99c7f0 : SUCCESS in 6m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/acf92fe3d99048638e8a2a294954960a : SUCCESS in 11m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f13613349a014157bca064c37ae0ea05 : SUCCESS in 5m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2671fba0feac42b5a9623a9059aee736 : SUCCESS in 5m 43s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fdc6252c542348018c08c4f5a6dc0ce0 : SUCCESS in 9m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/eebe7da2aaa0410ea207033fb3f20c5b : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dad54231fcfa4c94bbb0a6219a2ae4d9 : SUCCESS in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75b2f5aef91c489b96e437e86cc3a880 : SUCCESS in 8m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f7e4b12a315046149e7152f7f9da8f58 : SUCCESS in 14m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/774953dd769f4b69b416f77c14d6935b : SUCCESS in 4m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/324830b904654c9792564131fd4542a8 : SUCCESS in 12m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5747cd03156f443c8f1272b0fbb3f6ef : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a82c6fe39f974219952e863a4b741432 : SUCCESS in 25m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/098104928f394e599e54803acf147368 : SUCCESS in 14m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a0b29a84fb914d13ad4d82e947151ce2 : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67c31da631c04f5b9930e4a6f7ebe0b2 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7db301c988744a40bc7912439ddc635d : SUCCESS in 21m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9083a640b985412786c93a81c48f4b84 : SUCCESS in 6m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/49386dd66064455d9a39e80f20975abb : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c58656bdf1e4be596223703733b59ad : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5fa43e08a6294fb79aec173bb88960b0 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7768c2caef1748d4b6fb245ef6943a35 : SUCCESS in 45m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/560d3b1c4ac44107840af9100a78431d : SUCCESS in 1h 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"cd803161569b98575cae1c4af46bf1c1338f3369","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 00:20:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d225c99a3d664fb242b5cd7b0e051d0a51e9e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 00:21:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08e24aa8b442ad6cb44581314cf029d25af5bc0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 01:18:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2956a8f12051488db4a305d2f7cf77da : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838bdce809ea4717934ef53d60530743 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77205dc5f16e41c38659274079de0d1a : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83168a0fea534438abb98492db95ed1c : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d50e684916f2456fb3f3d1b41e81d9c9 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4753aa2e24d4434785bff9e0a0619e47 : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd2ee1094c93489c9dca3275428d45cf : SUCCESS in 10m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad5e849c810e4083ae7ec645dcb88509 : SUCCESS in 11m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f5f6d6f3e9294f55ba83aac193dc7e23 : SUCCESS in 9m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/988459866b5f408eb448b773b683588c : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b2a53ed2687b403aa03ef79d1300d713 : SUCCESS in 5m 51s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/de899edbcb924e94aa128289c241f2f9 : SUCCESS in 8m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d33c897819554f60bf158e485002b9ba : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f8e7d5a0d124df2bf61c719da43aa5f : SUCCESS in 8m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/792814c1a28540899d0fd3c7f9bb4d02 : SUCCESS in 7m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c5bd0b3a2ae6431fa8a2a3db6855cd4c : SUCCESS in 11m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0b172cecc3634de2b28645fe4b5161d8 : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/925c8dba525341eab9eb76d3c46c5675 : SUCCESS in 10m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2d71d0efd64d44a98047342bde94445a : SUCCESS in 4m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9cfe7f06d3a0496a85754b8664b35567 : SUCCESS in 22m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b8c7f11dda6d4dddade271fc8523223c : SUCCESS in 14m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/20e007bdb3a6488d924a131a2a7802e5 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd0f8845ed2f4b5f860befc5daa25725 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d3bdca40c604fd6b42b3faee2e1c101 : SUCCESS in 17m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ca7ce866a03c4363b19ae8e500e28499 : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/55ebacb8ad244a0583fde636110e6f1c : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c699a962a25a419ca7b18433482547ac : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cd955e69919547e4bd268f34ef835746 : TIMED_OUT in 41m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5974b99ea35465384608dde5c3e32c1 : POST_FAILURE in 9m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98b592791d8746339d6b04df4a9258e3 : FAILURE in 8m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b038451e6419d7e4897fb387d1d54f13a120fe4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 02:01:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd3948ecd7a12efbf8547d384057d01bb51058c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 03:10:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ecf6798f6764e62a63dc253cb3ec186 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d3224f58ec407f8c0c7321622d489a : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84d50574fafd42499eab227c6ce2abc9 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11f38f4e6cfe4e05a13be6625e4590a9 : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99df81de941f423e8c36c84d4ca32aeb : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9db2a4bca49746b2beb5e8507b8d6b91 : SUCCESS in 4m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e9f12bd73c054855ae753da449579419 : SUCCESS in 9m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0d809a72dd5641068a05d8d50088c02b : SUCCESS in 12m 32s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7551aeed353440ca998e112c50d3ba63 : SUCCESS in 8m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8497c2983a0e47429c3a19814693dffc : SUCCESS in 12m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/532d20a807d84ea3b08ef232ef334796 : SUCCESS in 6m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d8ebf30c629b4d5f92a986dd6ab513a8 : SUCCESS in 8m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/abe62b638fae43458d7a376e8f4965cc : SUCCESS in 9m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d90c923607514b5080c381edbb0d27af : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46c80cdb7ce7454bad1c7703da311a5d : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/60b7316b28c94de5bb1a239ce14d2981 : SUCCESS in 10m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3f53edf761864005bfee9406782f3862 : SUCCESS in 10m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bb3ea40182974041a782faf9dca6986f : SUCCESS in 6m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dcaf1c5e45bf4cfdbe2e216dbc39ed00 : SUCCESS in 9m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6fedbcd1872c4a15bad36502e4c637cc : SUCCESS in 6m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b49ad2cb01d044c39967437d3d5d640f : SUCCESS in 25m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cf12e7837e624bf394f8eedb48fce145 : SUCCESS in 14m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3f10199878c74ae78495750b3a03b21b : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90ceb3d3e85140f387217bfac8c2d069 : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d954d8c5fc894ac6af7a27777161dd81 : SUCCESS in 24m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3e77084678a94242890aace570326122 : SUCCESS in 7m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f915f02095c34b38b67b15799f3c57f7 : SUCCESS in 8m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38897fc23aa443ecbf2299a7f1a0ec75 : SUCCESS in 9m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ed0b5fe3baca4a0ab6fefecd03823baf : SUCCESS in 15m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b20841840954938a4b5e19ba25cd649 : POST_FAILURE in 19m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff25aed434974a7896c0c5deb52f7baf : FAILURE in 20m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e696181ae73e74e1876cc861c85835625c0aa938","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 03:42:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b3600d87ee7e4d84af28cb663c9ab8d6 : FAILURE in 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5178dbb4d1ffe8553b10f46d553c934ca10c634d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 04:39:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b7ea5261731f32b3023ab5c7593c1ebac1dff15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 04:56:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a21b28af55e94f49a1550c266c5b978d : FAILURE in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d62e12701273b890845c59624d2deff8a535cad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 05:10:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f7b12b51cb45ff92edef12bd740145 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf9f0e2f60949df96ff9adc474e685a : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ab114cabe1e4508824c9ce25cc227fa : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b35e86468be24da3beed95e35ed92908 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4bb68aa12594c97af948d104b035aa9 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/405e8633e54f40cf9d827e7b6e4b7d6a : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce417644a69d40b89139db12cb7f4315 : SUCCESS in 9m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8d8c89329954f06a7b7b6ab3e5cf9dc : SUCCESS in 11m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a8c7e1f6ffe74606a3f0d7a981148bb1 : SUCCESS in 5m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e7d4e7ddaea14a3dad4e16291971afdd : SUCCESS in 10m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/27c9bc789f99412b93d4722b4eeb76bc : SUCCESS in 5m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c13d7a2acaa04950b8efcdb40679d1fe : SUCCESS in 5m 59s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1b014fe68b014271a33258e626c2fccc : SUCCESS in 8m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ec1d93b014d646d0b6da997f13bba652 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8b46a8c85824371bdd9229d3bb2bc2e : SUCCESS in 8m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f659c3bd70b64045bc5cd8c7a211f7f8 : SUCCESS in 7m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/845200f396f640ae98c1a8a49e35802d : SUCCESS in 13m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3d57a6ba07ec4654af9a15b5095bc672 : SUCCESS in 4m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/209c6fb9cfb84560bdacd34bbb77b81c : SUCCESS in 9m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bd44f2386e7b495ea85b49f8cc1e6f82 : SUCCESS in 6m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f4402d09f96044989f9e45d3a0b9648d : SUCCESS in 24m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/90cd8c0cf5064b679a1e7a7af4cbb452 : SUCCESS in 12m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cfd2c45e22b344d3ae615b9f5c5aac03 : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/795e9bf3f8104e4cbee5259d8309adcf : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e72abcf9c6f44c8cb332d51f5179374d : SUCCESS in 18m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/469e3261b066414daf05cc64339404d5 : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/11e9fd7af6804027844de171437e12e1 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dfafdbe07dda4349aa24dcda50910e04 : SUCCESS in 8m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/40f4d7b44ece444c8be6b6ee3c0c51b1 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93d5254d2185414a8b710e8617c08cde : POST_FAILURE in 16m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/387b686cca0e4736b15c3e2e7bcb2d53 : FAILURE in 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bffcbf00876bc78f3d05090efbd14a461f785fbf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-13 06:50:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5078e90fa7b973cbd679495d0d7180968e4e1b3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:17:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/50d4394db7524a7fac3ff11a55bd3575 : FAILURE in 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f064c7082798b6a789cc066906ec88dede8c59fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:24:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7f2efebafd41029febf70ce3522c9b : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bdb2e477f54582b006f9260718d50e : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b925cf02bad547f9a16e89557bd260ac : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/043eeb884f9e4b378ad69613c7abbfa9 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf5a3ac15e0b464c92009e5a0b104455 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1b3dcbf80924074bfb89e847be3a9c0 : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18b972eb43a54be69218909131913fc7 : SUCCESS in 11m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/218365307a18445095e960edc5df39ed : SUCCESS in 10m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/aa3b0d2b5eab4a39880d0841e7d094fd : SUCCESS in 5m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/021fffbb321143dd94e632f662daf792 : SUCCESS in 11m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0b239f4b33ec42b9bc758b57b81c34d7 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/df6e6955974f4fee804f3a774a562125 : SUCCESS in 6m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f8f2af4183d34a58a41d6134b00ed077 : SUCCESS in 8m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/69cf520501f44675a56886afe5541abf : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/469d7f960b684a44930f2ef294d80378 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4bcfce4602db4eb182749509e5e92a8a : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1006436182b04a57a2ccb86d95d67b82 : SUCCESS in 23m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f9adc8c49289464d87e33e6fc1668a69 : SUCCESS in 4m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b119d4debbaa4f8b94962da3faf8483a : FAILURE in 17m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ce82d19cc08947af9c0ff91e5d805f33 : SUCCESS in 4m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e80927e9d6404885a11bd9b8d13d9a64 : SUCCESS in 27m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4c32ac3f665a4356961ab2fb1c79af68 : SUCCESS in 18m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/224ce2997a7a4da1a5871efa8c50bf6d : SUCCESS in 6m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d7830408fdb41669e4b1f28ebaa152a : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a759fbcf189449eb3377ef2166a6b84 : SUCCESS in 20m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8e418d8885ac467886675659edf57f3d : SUCCESS in 6m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ea7da3e90f3848178faa2abfc9e74879 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3352d282cbaa4fe7b1f3f30127965ae3 : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a1da2c2717d54d2f9ae828048dac70c9 : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76d5571aa6f54854b18cfaadb5f6f70c : POST_FAILURE in 17m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae80888c436644c7b671966019f4884d : FAILURE in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e8d76b7057baa9b21592afbfbb7d290756191259","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 15:01:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00a8d7d6699e43188e18c9281aafa35b1b36f8e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:21:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a8e7a1ec2851450fa299673782593206 : FAILURE in 17m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b16aa3635b27328a8b960b27cec35eb492b72d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:33:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e4ab112eea45efa01c23e91ba6a616 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d341b2035e0e44dc80fc4da3d9680ad8 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db7b2f72a9ac4bdcb9c54687cccdbe93 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a928d67d23e4dd0ae7224400d5a4554 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/862345c19d2449a3a9ee98ababfa2919 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f07ab8c4c98543bdabee8c05c1e29e6c : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a97f0b1fdbc44a8805ecdda0cce661e : SUCCESS in 10m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ea89f1d839643f8bb150a05599b2e58 : SUCCESS in 10m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/87dd995c978e477484389ef8e38986b1 : SUCCESS in 6m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/740b53d18ee94933a0d171ba714e2485 : SUCCESS in 11m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b17ccc69d26d46b4905ed89893bb39d3 : SUCCESS in 5m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ad7d0ce05e2341d69797051c4863543c : SUCCESS in 6m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/efe0c072345e42aea93f8345f40b1b55 : SUCCESS in 8m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0233f9b1909d45bf80b707b8c75f8ea4 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9d01f6ac7ce4604af01f0ee169a6584 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ba88824c2c51438c9730bd6a94b82888 : SUCCESS in 8m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/00825aeeb15d471a90abe064a97371da : SUCCESS in 11m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/100b7109fb1a454faf41446591d5c346 : SUCCESS in 5m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/76632b53d2524b16a7c12791337af638 : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/599846bf996b4098b4078e4aabe8f732 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9d88449e38c14d56b9eceb1af2e05143 : SUCCESS in 25m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b76251a3502b42cd834b2c33f4482830 : SUCCESS in 12m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/29a3a8a183dc46069df70f79644bd0e5 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/804ae0822a1745629122f3fe0a60133f : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ba4d1388fdf4132b711b142da9c905c : SUCCESS in 21m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/41c28939b5ff49be9b05cf528632047f : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/06128aee261141e48ff19d35c5ebd312 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/535b4d93025b4cdbb56b71222b09d861 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/141ac0da24b14b86b299efa6e32587ac : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/feff06e7114f4a7fbef36f606c43b1ea : POST_FAILURE in 18m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0a0f8aa994b43ba9dda125bcb67607c : FAILURE in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d637a437f630722cbe578bd9ed0d4d12f5125762","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-14 16:08:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8428b421742b23e60f185c4070af7668baf8d393","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 16:54:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c0c557af3bae4b5fb3a57322e82bf57d : FAILURE in 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"65da971b826c4623a16d89521387ad73841647fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:51:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1da2c8c83f44058b0155b7a59b87f4 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb52f5aed384b11af5d1ac1c510fa56 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ce9ace33e5485391735ce9d89a537d : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ded1d1b628204e7ab29185b9aa155c27 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47cb0bcb895c481d949d20cae2e39740 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e136f37f85d47dd81ca0c6321000306 : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/636631dd4ac44ea5920ab0066588fdf9 : SUCCESS in 11m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c06ad7627174a9ea76d8604d5691fae : SUCCESS in 23m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/eb0b4aad995644818e99487acf8e86d9 : SUCCESS in 5m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/67efc4d0957a4b43ac61eab99600bdd6 : SUCCESS in 11m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/73e729c60b2c4c4cb27355b1846c2253 : SUCCESS in 4m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c261d7c6f2664fe7829ba6d0d43cb974 : SUCCESS in 5m 59s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f649da53233e40f5b9b293caf5d58d4d : SUCCESS in 8m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/13a906a1706f410f9aea99fd47db14fd : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db1693631ba04a09873585c627f1b7fe : SUCCESS in 12m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5eba09e412d34d90815db023165fd144 : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/86302c5172e344308b601376aa99fb93 : SUCCESS in 11m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/80d1aa8998c542438ecdbb0fbc6a8f61 : SUCCESS in 5m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a9438ba2cd994f1786c773eed8726c3f : SUCCESS in 9m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/141d90e5b4574ae5a0a0de036d69d68d : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e9202dd0d7494f49826b354dc6d35b53 : SUCCESS in 28m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b4b5027655dd46008ac308b2e13a7e4a : SUCCESS in 12m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/43ab3ffcc1984c65a23b60bda47765bf : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e644af2c0643482b81a4cf6f7946a314 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87eee1f6e8774cebb3de2024d2eb015f : SUCCESS in 20m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b9ad1311d1d0450f91b5353dccee7587 : SUCCESS in 6m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ddcdfcd09e154c3cb9707dd74c325ad5 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3d58b7127b64abe9e9050a291ed7a4a : SUCCESS in 8m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7e2a3e09d7374cd3bd57a1dbcd87755b : SUCCESS in 11m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175df5a7d3334a53819d7d01b7541c6f : SUCCESS in 4m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5caad5172dff4c6bb311b68849d9c283 : SUCCESS in 49m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a83696075c5743bfadae8f921cff2597 : FAILURE in 1h 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0ec5a65a356c270a65c8ef764f95dded7869eadc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-14 21:33:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"268013beb74fae038363590e47e6dc301a10f515","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 21:50:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5bbf61505ad44b918a073b256bf62b52 : FAILURE in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"119bd6fcf6ad40b74f5e1a86086259eef824aa02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 23:29:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b5add65f3b434aa453958a6469989b : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8f23016fe14fd6b1e6858d9a7b6ccb : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78641433a9c84ad2bb0c2aecd6dc4f74 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f10dbfc59b9949788af71d10bcfd34b4 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2733028f0894532874b775e29d7b7fd : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a08338075df4d43a13093a644e300b3 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f5238fb32cf44008af9337a8194fdbb4 : SUCCESS in 10m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c1a8fe08befe40f3a0ebf5b413703874 : SUCCESS in 17m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/18658a81e4c049878263d4711464586c : SUCCESS in 5m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/92309f20e0bf4002b2a419042648715c : SUCCESS in 12m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0257f6a0fdd343bca656d422f8c6a0fc : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c7116ed55f864bc085f3ece021f62674 : SUCCESS in 5m 40s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d4e68b3728f54cf5a12da1ceadb40dee : SUCCESS in 8m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/186bad05a59c4e49a47fa6446000a35c : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58fdb59d00e94401a3651d08696a753f : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3653950d4cd14ab1a05dce65b8aad24d : SUCCESS in 8m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7be593835d0f4a89b301e66b0035e71f : SUCCESS in 10m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/be9c663d99a448e98f78ced42d5a2e8d : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/561ca17d957547ff9f05f6906f8617db : SUCCESS in 8m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d595786bf3534eec88af2d6c4d590363 : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a9b89721d53b42e985d5133144a5878c : SUCCESS in 26m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3facb7474e004ad380664d3c236f6188 : SUCCESS in 15m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/928205641eb64b9cba94c82dd77c51da : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48c2f85907f74eff8d6211a1bdeaa59b : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1689d1c32ef48a3af30fc873fa5e1de : SUCCESS in 22m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/eb47d27149054f5baaf944aa95e782c8 : SUCCESS in 6m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cfc0870303084b169b7f8fc322a359bf : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9555369d2a2d41f089655b04cbc76ecd : SUCCESS in 6m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cbe9ec29aaa4454fb6dfd2e8b8c89bc2 : SUCCESS in 13m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/903f6c343e5241239a1449712857816d : SUCCESS in 4m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c644c90888674ee19d4324da2f0c6c82 : SUCCESS in 38m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb8f5fedd29245a98bb44358216c9807 : SUCCESS in 1h 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"69aee037f8b725f4b962578b436cc34511cab801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 23:30:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"798de1550ab7c586a555855a199ebc742b9af497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 01:09:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35926e09fc65413cac12804d87cd166b : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853d9d4639534632b465a2d8e21f666f : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4ef8e88a7a24fe3be5b92d7a8617855 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93e348d056b44b418f897da76c061b6d : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c71e3bfce91a4bdda48696da3d77a51c : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38d5a2cbec1d4eb4a1f8504b6717abbe : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c847bf37751944cdbcca192864adf852 : SUCCESS in 10m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/89c0023daaf24061bd7d8d7ee0a22205 : SUCCESS in 18m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f32cde052ba34b7dac2eb320157d7a3b : SUCCESS in 11m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4337c9b2478c484683dd723f57dcba1b : SUCCESS in 5m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/678c097b1508496b85274956e1f88eef : SUCCESS in 5m 05s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7af7e11d633a43e092df00e6bb51e94e : SUCCESS in 7m 30s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0993a6d2a29f4fe3b0e392cece1fd362 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/677688ff7308439d9730865921790b8c : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/001407d3d9174e9aa14173bce7be4491 : SUCCESS in 7m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f8419d4290464e00882db21ea9a7c7e7 : SUCCESS in 10m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/35cce602bcf74444bec31ec725904953 : SUCCESS in 4m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d3d180c22391461b9930c7f9675157fe : SUCCESS in 10m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/479e0c9aadbe48fdbd2d3a535f3d7bf2 : SUCCESS in 5m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/df0e3f835b95407aa7f077d9eac1d24d : SUCCESS in 22m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8b28c32bde164741895ffc3e83a9f003 : SUCCESS in 13m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fa9c7c6890244d2f9e21a6173a1c379a : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/996d5c7aa15d46c1b96b8f39675ba8d5 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee42a2a09e134ec4a7cf71acdd165a79 : SUCCESS in 19m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9d040c545d8b4b2c8a675bf85b72fdc5 : SUCCESS in 5m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6365e023b95b480a9627df13723ddc0e : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d231294e05544a38b4481194873701c : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1a48e672e6c54343be22a574c67543df : SUCCESS in 12m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5721be8721447f59bc9777298b5a06f : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e20b16a36fef4fe7b2f417413928aa45 : SUCCESS in 36m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06d92863e3d34dd0ad2f8ff37c9c657f : SUCCESS in 1h 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fba62d050f8e7047554d6227758d0434b9179dff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 01:09:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e1aeb0e594cc4eb32f22fa5fdfcb62ab303d786","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 01:10:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf6dca5cdb146a082d8e1bbac5b52ed : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eea574e29ec12176465914ef5fb76ea6c6cd0bac","revisions":{"eea574e29ec12176465914ef5fb76ea6c6cd0bac":{"kind":"REWORK","_number":1,"created":"2022-04-12 16:39:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/837598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/837598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/837598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/837598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/837598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/837598/1"}}},"commit":{"parents":[{"commit":"7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c","subject":"update constraint for ovsdbapp to new release 1.16.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 16:39:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 16:39:31.000000000","tz":0},"subject":"update constraint for os-net-config to new release 15.2.0","message":"update constraint for os-net-config to new release 15.2.0\n\nmeta: version: 15.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I8f55b4f4011ee351b0f2eedff8e5d484a9592182\nmeta: release:Code-Review+1: James Slagle \u003cjslagle@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5b7096e38e7c2d6a567528779d71df7fe504809f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eea574e29ec12176465914ef5fb76ea6c6cd0bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eea574e29ec12176465914ef5fb76ea6c6cd0bac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
