)]}'
{"id":"openstack%2Frequirements~837601","triplet_id":"openstack%2Frequirements~master~I7041eeabd143f84fb04cafda6d74b6c9b1d9c4cc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7041eeabd143f84fb04cafda6d74b6c9b1d9c4cc","subject":"update constraint for validations-libs to new release 1.7.0","status":"MERGED","created":"2022-04-12 16:56:28.000000000","updated":"2022-04-14 19:30:24.000000000","submitted":"2022-04-14 19:28:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837601-new-release","meta_rev_id":"708a23693e439234a3689bf854559759037a7808","_number":837601,"virtual_id_number":837601,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 19:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-12 18:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-04-13 00:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-04-13 06:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-13 01:05:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-04-13 00:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 17:17:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-12 18:13:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-12 20:21:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-13 00:20:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-13 01:05:38.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-04-13 06:51:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"35f92e7dd821683ff0d4b25ebaaa4a8a4eb8c422","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-12 16:56:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfe1eb1b074e308df2bd217cb4619efd7607e7e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 17:17:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4804226054ba4814b7b59ee4cd4b342e : FAILURE in 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"affe610da2ca94f16ed7128f4630113d3bf4bfd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-12 18:13:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d3c1c33211e2be476c9c806adf590b74b8b3274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 20:21:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7262260b27ab439292efdb41ebf12e24 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7ea2edd646a4d6a90a85f6cffcc3ea8 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef86e039ca534d7ca767bbacd9bc06d7 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f9603c653e24a4fb67f7623a57140d4 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a16c3b5a7c364848b8769612fab6ce5b : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf1c5f5d2bf04e838a903b5f5b1190d0 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5afaf7d3ae4744fbb9edf1fbf2c31ce5 : SUCCESS in 10m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3820f8bdefee4a2993d84267825c8a8c : SUCCESS in 11m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8f1440a531d8416bafc357a8bb68dda2 : SUCCESS in 6m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f8d81c3df2524c4093ae1ab52805c3df : SUCCESS in 11m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5b5de02f59a349c8a84a85a2c5799dcf : SUCCESS in 5m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/81defe84090944a8afc270d1f4ae3279 : SUCCESS in 5m 59s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fffd3c9ff7bc43e387c33705d6d78b28 : SUCCESS in 9m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4c7e071d6f144dbb946c15dc3ade5d69 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5da66b68a8054f7f8abd5760145509c0 : SUCCESS in 9m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dd7ca2932255482ebf1b54e5484a9315 : SUCCESS in 8m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bc6c8f37041847459a738696482cb1ab : SUCCESS in 13m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/501f8b86ac6f420298d98e340b4c6888 : SUCCESS in 5m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/25239acca2fc43f5a243f0c8bf2a40e7 : SUCCESS in 10m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6ba02d3d5f96404bbb0820e902e85ab6 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2f526afd10e34aff85caa2de23949757 : SUCCESS in 21m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4cd6a372b4f546f68a2d9ea092d48078 : SUCCESS in 13m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f760c8cd8cc943a5b5dd57e6a7ee86da : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a86464d6d1640928f3cb864e0b3fdf8 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d2239a0a65e4ad0adf8a8ffaa57f5d8 : SUCCESS in 19m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/343482b4446f47438c84ff0966eaf62a : SUCCESS in 7m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8b4561d08fd54c0b8df30ea00cad27f3 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0994cdcbe3984cbeb91d3cf0f0693d21 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/829139f04b794715b36368f13a2b7bd6 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73af7ad6b032415ea900e3ab7411afaa : SUCCESS in 49m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f17d3474b9643cdbe0900290df7122f : SUCCESS in 1h 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c4970513ae9a27322b7a6a3b0ecd38e7aa27e41c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 00:20:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8d244115fa4d390d46cdbface616657afdfcce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 00:21:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70bcc0051c6e6001ac990252b2bfa2d4d412eb85","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-04-13 01:05:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1473d99de9d3c7425e37bf8bfeb60843c13c8229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 01:39:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef4b1b44d2641abac4a647c1327db2e : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e35e7e98b39c4fa098e919258995fd42 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47cf6880dd594355bd5f6c12e6354b0f : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5accb29dc206486393f01ef48b5da93f : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68974abffeda413eac9090d843345402 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d8ebaf097994bc08e0f93a32eec61a0 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb15044022dd4958ad5c8f6798a72f77 : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a842ebc290574604a9039f00fcf41a62 : SUCCESS in 9m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6dc5fa65789e4f0dbfc6b36363a745e4 : SUCCESS in 10m 53s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f9ef4423fb6b4cc197634d5e62ff38dd : SUCCESS in 5m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1e55a7c79fd5481c8161088a1c1f8970 : SUCCESS in 5m 29s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/750e20a3ce1a47e9bfec3f96958f8084 : SUCCESS in 8m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/14d7e31e6da949f1b33e3c2b8728a19a : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/573c890811674aae973aaa1da3fc850c : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/53f5b555f766417686b577e4718d6c4d : SUCCESS in 7m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b3d53b2007024b14b647a6b68de61757 : SUCCESS in 10m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0a46f5d42eac4ffc9d2f7bfdbfd8258b : SUCCESS in 4m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4a031a2023b44e5fbc17e7aa0f859a44 : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b985c012e2e945438d652475810834fb : SUCCESS in 4m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f63e8b844c3146cbab91b8c6f6aa5381 : SUCCESS in 21m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3f8211961a9449ea904ea288ecbe0476 : SUCCESS in 10m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c1e0532b9e6a4065af4bf9a7e6738bb9 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c16461ea8fe84342abd4d7615cd31d79 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7ab1d57cda04961b9c7a34c8b846c85 : SUCCESS in 18m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9327c7d7ef3d40fa9e1165d74b7c41aa : SUCCESS in 6m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/21b181a893e740a597ca60a5c1408bb0 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c585d3a74aa461ca16712a7767e1a84 : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/629f9f5abdc741f79b598d2a105bd66e : TIMED_OUT in 41m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7652429f6221493cb95eb4172cc9913e : POST_FAILURE in 9m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/783b85da7b864b5fb6141a1f3046cac4 : FAILURE in 9m 02s","accounts_in_message":[],"_revision_number":1},{"id":"852e6609ec3832047e4fd629568cf4dde9540c73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 02:01:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79fa4dfd1c4dd6900064a63404bac2079f8f6594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 03:43:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac16cb349f9c45a588a2ca0d6240b813 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614467b1bce045948d04b3af7af0b7cb : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/510a79f1f5c645f697eb6a7c5bf916b2 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ce451a167924127b26a9dd936e96208 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbb3fc9adf97478685d8c1352c39716d : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c733d88c95f4651958b5723f1f7d259 : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1140d940d00c4c8faca82f5989a64643 : SUCCESS in 12m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63fe1310c1684a89a2f9f58669599c31 : SUCCESS in 11m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f826e9ee874b4ee895dd0732ea366635 : SUCCESS in 7m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d5a8c96bb44b4a94bf0d168523092d6f : SUCCESS in 12m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/824a18a3672f43ffb58a64c29a200410 : SUCCESS in 5m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/326c01c3cb1747f18baa70136b705154 : SUCCESS in 7m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c4ed91e373e9401189c3b7146f483c32 : SUCCESS in 8m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/87c6e1f2353f423e856ba67292d6f702 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73c0a51089d74d18868a50427d3d4a07 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6b42f2239d6d498195a00c60488a78fc : SUCCESS in 10m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c6c18423401a4003906ae873f06f0a22 : SUCCESS in 12m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ae1ad161dc614be0ba81830a5b714496 : SUCCESS in 5m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1c19db469675420183a165b80a264312 : SUCCESS in 9m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b96333bcbdbf465f85dec92caae85445 : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2df99320aa7f43f2a353e4a65b807081 : SUCCESS in 29m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9685d7ac7a88445eb4b57608a025ad7b : SUCCESS in 11m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/63d30398c24f49798ac7cbfcb845e480 : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6635373439e14df38fff94a567205318 : SUCCESS in 5m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e46c32493a73482baa46a278b71c6ba0 : SUCCESS in 21m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/af12cb7201254009b4f1c1864ebbec84 : SUCCESS in 8m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/597e1d6198e748e183bc4c362faea9b3 : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a8f2d37d6874230b85083a111ed28ba : SUCCESS in 8m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0d768560e519429d9e1ddcea1afec6a7 : TIMED_OUT in 41m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03b9984d62774d9fa0b5e37650c68d21 : POST_FAILURE in 17m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5f030c775794c37b785c2e08cb4b139 : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f4f3944ecec82e8f18913a8ac87304770a2ec751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 03:58:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3e6d004599db4ad88519baeb0321d679 : FAILURE in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7e4179afe29e37e898f6dfb06eb8a8597272f700","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 04:39:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a56d07a1a0195b90aae05d47a4ba09966ba1b1b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 04:55:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b5d8d1ce7fe24443bfedd3dc50e897d4 : FAILURE in 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cb9a6ec64e42e04a4d68ad31ad6b730232bc3626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 05:09:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1aa0451c894adca61ff47d53b53851 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d267f28dfb41af8756cdf4f066f7a2 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eef71428d4a14e959bd95c0d3a028fe4 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0b872742e334067a5ee5279672e2462 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/536e7f8dd04543c68d72fef3aa003a3c : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7ff922cca654c94bbff52ff40f20f58 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b181ac405cc4745a1985c415d53d0df : SUCCESS in 9m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f85eec6fa4a647419fe80db54186adcf : SUCCESS in 10m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4d9c95f55bc14b93a8dcf9d9b25b78ca : SUCCESS in 5m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b7d82b3bc60b4de48b145a65d7befd52 : SUCCESS in 10m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fa88e782a0894dd980552d1c18a428ea : SUCCESS in 5m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4504cf2c73a84e809ea0564ad9dc0756 : SUCCESS in 5m 17s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/cd3016307b8648dcac63bacbf16037c8 : SUCCESS in 8m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/698bdd0499c24d30bd154063b38f16b5 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/294dadc721784f9eb29811125f210166 : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d577867ee5f94805ba20ef08909d2590 : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/584523c1d8e04f5da53b9682e6f5a61d : SUCCESS in 16m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/84bca1b4a1a242b88c4851722727ab0c : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f539e1c4ab444e73a593cbc602b2f021 : SUCCESS in 15m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2984a2243b5c4bca94ab18ce79cdd4b0 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/03da645e18664577992387471861c4ff : SUCCESS in 22m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a906e7c77ca94d938785f6dac4936ce7 : SUCCESS in 14m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4d695a5a7a574684abb0e53e16a61671 : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56bce79412204828bb339d0772725c48 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54954350d4ad4ffb8eeb94a79a9e1099 : SUCCESS in 23m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5a37d77ce98447cbbcf09e512103267c : SUCCESS in 5m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5e1348b659b94ab3a5f7705eb8250400 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ac5c6799d674489a58c49d90e3e22bf : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d8737f66eead45fdaf490d336fc457af : SUCCESS in 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5da9f93299c4bc38fc1361c12ddc471 : POST_FAILURE in 9m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dce4bcfd8b24a22b9d33317d10903c3 : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f029eb6c61aa8ab7ca53c9c70c9a961c052d2805","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 05:53:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b566598f6a92d64e8ec7e71ebd786d9eb22fa2fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 06:09:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/db4214f6d46a42aaa0519d7755b606a6 : FAILURE in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9aff02e625afcdeccb0df0b9242dc5ace81872b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 06:27:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e359ae8bb742ea9adcc2cc7190ccf2 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b991acd7f345a1b168a6a9290b339c : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdca61903149453292625b4346516237 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53417eccfd7b4126809a1936033b66ea : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7aa85d000d494b33afc50898b89114b6 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5717cbd2620146d09e37bd99261b49cc : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b13bdf995fd04bf68a1ffa9659b1dcbd : SUCCESS in 8m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fa8d502a57454151a2bb11d00578b597 : SUCCESS in 10m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f946045b03f84c5c979e3640f40bc000 : SUCCESS in 5m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e69d0c727dc64544ae3ede0d6030ef69 : SUCCESS in 11m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8fa06037245a4767942797e0954fd3bf : SUCCESS in 5m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/875f373090ca481188ffa2203dbfe2e0 : SUCCESS in 6m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5328c0b5b91f4c69a8f60902ea20d637 : SUCCESS in 8m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/da5e0a02bb824d8c8f97198ae93cbb0c : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3bd440b73c4408d96c2f52a2a9c37c8 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07ab4285d4db469fa360909637d7100b : SUCCESS in 8m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ea9003a47a154cdb88690560901381bf : SUCCESS in 12m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7ac2f6bf109e42368be68c43136be589 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9382348a276e47298c139ca2aa39dc72 : SUCCESS in 10m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4e138ab09d754e4e817ef33a711dc4dd : SUCCESS in 5m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cb5fda881fbb4f5d9561b853ea5844ca : SUCCESS in 28m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/29cc26a48c794cf1b229a5ae8bc79d0a : SUCCESS in 12m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ddcf290c3f204a95a6fa50c186134df2 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/df100a11cfbd451eb126a349db07b99a : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8241fa4901d4e3ca122772ec772ad82 : SUCCESS in 20m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d8a9f67d68d34c4089780f7d14b086bb : SUCCESS in 5m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/51762cf9389146e981b1c8d2add4089a : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/667dac173932433cabe2b25fa6f0db99 : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/52ba4eabdccd494a95c757219834b7f7 : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ae143e9efe241929021523630808ad6 : POST_FAILURE in 17m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/426d17a88f4d4aa2ae7295fef7cc16f1 : FAILURE in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e790788990c6ab0129ac4c5082b05d3f665c9de1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-13 06:51:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a9f7a62f1ebc7013ea199be3e569d1078966296","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:22:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e04ecf4bfb6d4e919e6851f2c280f611 : FAILURE in 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"90f34c4edbf935bd97e1eb0086f5f2dfa5be992b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:23:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f63cc301fc4de8acaa99c9929033a2 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9e8372a1294f6fb55520d27cf533ac : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0606ca4d99974439bda3f9dcb2c60583 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46526d1c5e2d48b7b5a865a9130f1472 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65c49e31f7624da69097a46b46418856 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c3b3807470fc4178bf10fd404fdd8023 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/802c3f872ab04ae18403851d23a9f966 : SUCCESS in 10m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce514f72cddf48a08829892cab92a8a8 : SUCCESS in 11m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/322ff71af804440d8aa17999a885eee1 : SUCCESS in 5m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1abd5fc5c6814f208a5788f58dc3b11b : SUCCESS in 10m 41s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b544330b0a0e4aa38a5f4e27bbc2ab1f : SUCCESS in 5m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d0e2915870504096b91deb18e536c918 : SUCCESS in 5m 09s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b3a9fed4c6924f0fa95f18a39e3a659e : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/20c09658a4f641b19eb56c2d36d3e781 : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d26b9bb8e6c4f5196a0076d9208e43b : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fb61a4b6e2b741ac8566560e640fb05f : SUCCESS in 9m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d2b4afa00b5940499507799e10a4dd14 : SUCCESS in 10m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b0c22a2f7d9a43f38341bf4bd2159338 : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2efea1554b3940b483668afe78259058 : SUCCESS in 9m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bc292201c338458488aeb42179ec2dc5 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6a330a55cdd948ad8364f8d35ffec4d0 : SUCCESS in 24m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/05ffc7ed61534d57b7374e4cbc604ffd : SUCCESS in 12m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1c12d403d5e84cf59174b1ac54f82233 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d555c58232374fc988ec0ab5d0d995a1 : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ac3c56771c94b158bbec09d15c2cab2 : SUCCESS in 18m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/77f804d122324654b619b9e98452b444 : SUCCESS in 5m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/510126e758f7494badb023cc3ae5360d : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48cb11a5db2b49a1a8fa6e262a32c5b5 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5779be344d0440648eb4fcb3343dcfb3 : SUCCESS in 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e34dbb2112174319b247dbd2d6b3b56b : POST_FAILURE in 18m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f9e48b6b97d49d6b182f51ab6084c99 : FAILURE in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"05a1a981921c155e43a68abdc576835fd4e309a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-13 15:01:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb934ac899d006d989d91fd40be7a2cf1a230958","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:20:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0cd56f7928794923b959fe2466bd885c : FAILURE in 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e04db5d44e2f5cdc105e8c53d6b574bcdf5348d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:32:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b81dbb432ef46318c1d0a5a3bc28863 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb41972829c41c6ad347a1a1fb48e7d : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f7c190f90a4efdae1183cbc521e30f : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8a1f7eea89a4bf1aad5b865ae3bdf3f : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed8f8ff15ef5472489375f53038b99b8 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b1b3f088362491a9816f218cee630f9 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cac56ed0d75b447c8e034354666f7daa : SUCCESS in 9m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8bff6cdc4fc64afe830c3862ab49a673 : SUCCESS in 10m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cc9d7802c3c94170a9abedcb2b3ab402 : SUCCESS in 5m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/16e5c2f385e9419d968c2a2952b77c95 : SUCCESS in 12m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/27a722db7dcc4ae9b18cbde13d675a3a : SUCCESS in 5m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9996c768a53940adbad4f91985e7c76c : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/10b1100eb9454b00bd824c8bb88c8595 : SUCCESS in 8m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a5b29ff553f0447884eee0f3338a43ae : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df64cd88d5794c1e855948b29217183b : SUCCESS in 9m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca86e13e11a2436d8e765e34ada850ad : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b0119e570b204ccd9fca9869dda6db4a : SUCCESS in 14m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e91075bb9ae340cbbfc9b51507ade38d : SUCCESS in 5m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b1f7e32d76834589a13b7224021c8d00 : SUCCESS in 10m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2cf0962ebec145d4a28a4e8569bc2b2e : SUCCESS in 6m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bb048c4df00a4876b3e49fd0d8b39148 : SUCCESS in 23m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c2a1c74bf00f491babd937c1cb0939ad : SUCCESS in 15m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cc182db759ef4b01a8d2bd55a0bde535 : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aed6344bcc5d4b4eb5eba5d50b4eacd4 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55cd5f9aba7f44cfa16a5909ab748daa : SUCCESS in 18m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e2128a0a583b4ecd9f9cc5c9491cb338 : SUCCESS in 7m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b59b0066870f43e396ece6107b6089a0 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3744203715a486a83897bfb6296bc9d : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dcc66d6e54664a3883bbcb866eb1e21f : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e407f6fbb5e745a19fe126bec36570f3 : POST_FAILURE in 10m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9135be951de2428a9140ab2cd15be580 : FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"01c6d1231a3dd3d01d3b9c668dad4e68487752eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-14 16:08:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c59b3f84b7c1a9c0ae565c711d6c51da180714cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:38:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/31ddb011081f4bf9beea65ed35c6081f : FAILURE in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"04d2d508958418d190300609f053cdbc42c534dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:39:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b170b6e41c461f8f245ef9e2a85121 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6be292c092e4bf7b9a63ca46643e9dc : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82c31d955eb5403e90db9c1be12bd5d0 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d59a38c73ac47919af8df539c44d731 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/625e3bb8a38946e0aa09e5cc51173ab8 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14bbebc112ed451293f4470b63765640 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f52eed0033bb48678a5fd9f6ee76ec10 : SUCCESS in 11m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f2ccb99758b544e3b385590997f3d57e : SUCCESS in 22m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e95bf9c16b114e6aa49c258c24f0279b : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/802172feba2449c5b3271b99af56857d : SUCCESS in 10m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d8880dc31ac94ce997309039a2fd4659 : SUCCESS in 5m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bd97dfa8def94fce852908f9041e2cde : SUCCESS in 5m 55s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/df5295c850f14245ac110c39da50afb1 : SUCCESS in 8m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e1a14b3e15eb42129e7a024947f4e63d : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e022ebaa11d64f8282379bd2c26f2f77 : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eb1b46211c21490ca73dcdc2c0982863 : SUCCESS in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ff784432a9714315a049f99ae9ee6685 : SUCCESS in 11m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/486bbdfdeb76434b8b8c2ba3188247d6 : SUCCESS in 5m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5200a0f951304a7c8cdfcf579ac8295e : SUCCESS in 10m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e7954c3d517f48c5b333db0462f8ebe6 : SUCCESS in 5m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/acf7dae7c2054b1bbb70c5369f122b37 : SUCCESS in 24m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2c63c7b5c9fc4d5590ae0dd65ccccdbb : SUCCESS in 13m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/afb0b0d67a1b4de29e21f48860ecd99b : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/65578804534c483ab5e308eda75970a6 : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4edb7681b3c647f88f142c6c25bdeb9c : SUCCESS in 19m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/980ab7d01ecb4365ab24b57f5d2ce6c3 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/98f013ae36014efdbb0497d1180af2b6 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94cf5a3d85be438db8043ce7b9bd4e53 : SUCCESS in 8m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8c71585fbca94863a81785acd5d1093f : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54e0ba73eb5b4f9282339145b78c4c64 : SUCCESS in 4m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1801a93933b64156bc981968b9b63ebf : SUCCESS in 50m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52782ff056d047b2a3bb60a65c196120 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fcf5c5ce3635ff148d6a56bd4b89eaa40ba2c457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:40:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae6965bf16171427249091d8dc6b39d8ebccb60d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 19:28:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05236b650dc40b2ab9a0e1f96e16071 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb5d72af9c014f68a0e98dd17ad67bd4 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d27d863624fc4b9c881adda81af0fea6 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21d195ebf6eb44659f0ff0496f019a61 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/def96523bc564ed88e49ffc920599dfa : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9bdf0b97e8b4450da34203d160c47384 : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/020489e43d6546668a24e52ea0b1a2e4 : SUCCESS in 10m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/74c6723f5f194e0eba588fac9a2e8167 : SUCCESS in 22m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/78ad52f6fbb148b2b174d5df0dec0c56 : SUCCESS in 10m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1857ab9f19c345baaa756df198128033 : SUCCESS in 4m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7b0cbef0cf5f4e74b4fabd03edd6989c : SUCCESS in 5m 23s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/cb3095bdca424e90a353014919bec984 : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1dba1059302840cabdbf392aac422d70 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0e1da3695dc433095d375746a8ade98 : SUCCESS in 9m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e51df25710bc43e5bdc383f332c52555 : SUCCESS in 9m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a0b88a10fb2a46679288e9c6802d4ffe : SUCCESS in 11m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f4529e242e6849269b1496b7360b6054 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/38fc62be30c243bc9e5f8a62ebde5071 : SUCCESS in 10m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d5c5da07d7604eedb99619411ef09fc9 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7ce0f092cb7e49a890a1bdc64ad94306 : SUCCESS in 27m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/464775dd0a6e47aea13fc71e12df562e : SUCCESS in 12m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4a8c96731aa744e996df84a0b498df3c : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95ddff8dc5d54275a736957577403b01 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/356b637df8c44d3a8dcc1a4431939e55 : SUCCESS in 20m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/205e05be0c7143b98fb1555afcc1e608 : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dbbb0e6002d04b4d8beb51f67f2da1c4 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d137fc05858f4d1086e8685a68a4a5be : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/33353889c1da4b6b94deb090138516ef : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf76c049f2e4c6fa668e1636be373fa : SUCCESS in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fee1309b245945bca41b00cb8c6846c2 : SUCCESS in 1h 05m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91e9a726c8584e768bf08523c8609e6b : SUCCESS in 1h 42m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0e44cc88f0083279444e7ee6897ecfb35a4c6aa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 19:28:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"708a23693e439234a3689bf854559759037a7808","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 19:30:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0e2a40afa244d7a811f88856fee810 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd461034a114f1f01f3c97fe8000b0b5c5e2e8ba","revisions":{"fd461034a114f1f01f3c97fe8000b0b5c5e2e8ba":{"kind":"REWORK","_number":1,"created":"2022-04-12 16:56:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/837601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/837601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/837601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/837601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/837601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/837601/1"}}},"commit":{"parents":[{"commit":"7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c","subject":"update constraint for ovsdbapp to new release 1.16.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b88d6e8c57997919bcdf2b0ec27e5f8347ddd9c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 16:56:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-12 16:56:25.000000000","tz":0},"subject":"update constraint for validations-libs to new release 1.7.0","message":"update constraint for validations-libs to new release 1.7.0\n\nmeta: version: 1.7.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I8f55b4f4011ee351b0f2eedff8e5d484a9592182\nmeta: release:Code-Review+1: James Slagle \u003cjslagle@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I7041eeabd143f84fb04cafda6d74b6c9b1d9c4cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd461034a114f1f01f3c97fe8000b0b5c5e2e8ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd461034a114f1f01f3c97fe8000b0b5c5e2e8ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
