)]}'
{"id":"openstack%2Frequirements~838735","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I38b8bbca61f519929ac8625360ea4f10ca75974d","project":"openstack/requirements","branch":"stable/wallaby","topic":"cve-2022-24765","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-04-20 15:06:17.000000000","reason":"Maksim Malchuk replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-04-20 18:56:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I38b8bbca61f519929ac8625360ea4f10ca75974d","subject":"Bump diskimage-builder version due CVE-2022-24765","status":"MERGED","created":"2022-04-20 14:41:56.000000000","updated":"2022-04-20 18:56:55.000000000","submitted":"2022-04-20 18:56:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838735-cve-2022-24765","meta_rev_id":"0d826a9c65940656e4b0162b1ff6c8bb735ee10a","_number":838735,"virtual_id_number":838735,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 18:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-20 16:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-04-20 15:06:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-20 16:48:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 14:41:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-04-20 15:00:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-20 15:05:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2022-04-20 15:06:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-04-20 16:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-20 16:48:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9c5a10936f1616a92b466789126caa8c0a2868f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-20 14:41:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b1cbf8cccac67532209f6d782a1c72bd4496eb2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-20 14:55:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4c27e329f423b0b7f01ec0fbe372fe500aa7b048","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-20 14:57:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d93a12fc36a0f3be41b64137015f962ad43bd2ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:00:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/590e0a0571174efeb3d84d813544232f : FAILURE in 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4eb2ae5f41786a105509471452fc09a337c969b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-20 15:05:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21df3632632b0bb8726eeed91b9e619bf2cf4dc1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-20 15:05:48.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"43887fd03b2e7c2f3e35ef022c3c38c618d41eb9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-20 15:06:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8ffa83838129e2baeaed4a2e45fe8295daca3052","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:29:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2803aafd924c4843b0f3abab843a26c4 : FAILURE in 18m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c1f1c91a7d6f17e557633299c862df713b68191f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-04-20 16:28:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed9f587edf4e9ff3dff8dd006aad31331fa74ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:33:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c6bca4573a458d8de37858a6e9f888 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cedf8d4276fc48c29e5d97662602d82b : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf682f8a630349ef9fe40095ae39eb8d : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49379d7821cf40dda1c2fe875ca95a7e : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33ba0e352bc24c1eae40b353eb7fe1f0 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f38e77241bb44ddaae42435fbc4a44bc : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1441427367634ea69087628f1acebaf6 : SUCCESS in 11m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b55a85fe36404468b7c52de25a835039 : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a072fe81048d4b9b97dd880dcc672497 : SUCCESS in 11m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bd8833e3a0e3421690f60791fe5cf744 : SUCCESS in 7m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ffc0a526490c44f1b56f4e1431c9517b : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dceb5df024454ee596574b47690159f6 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0247499d9224445aa94c7c98a7f1cfe6 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/137e8c62f78c457abf7e3cdcbec8c41e : SUCCESS in 14m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2c4b3dc8ed02419a9fce8d9e205a55eb : SUCCESS in 5m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d109c780beac4c0b9e86aaccb575c859 : SUCCESS in 36m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/49611d9d7451432983e286c38d0127e3 : SUCCESS in 13m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2fbd1f8385f47b6bf2fe4bd5d2bf1fb : SUCCESS in 5m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/149187a2a48c46959d6733eeeccc1583 : SUCCESS in 19m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b514fb88c674e949c09860a18525409 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/433c9d5763a34253a3e801b4866d2f74 : SUCCESS in 7m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c046f4d24e64dc3a8e4d9cf87af099a : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c4af5ebf39d4100820643d06dbce8cf : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b70333857c345bfb35d6a772269233b : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5c93e1ccd4c8fdb6463361b8175056061978aeb7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-20 16:48:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0d9e89d52b455e6b31d8ddcddd33655845bb26d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:50:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7b291cf1285a7062f3393d6a344595baabf2003c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-20 17:08:47.000000000","message":"Topic set to cve-2022-24765","accounts_in_message":[],"_revision_number":2},{"id":"0bec83b887565de8f6957301ae97a8c5986b94a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 18:56:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12ad6f531464b52b66bd8317447a420 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf55a5567dd54ce0b04d6f005c3fdbc0 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a49e854d4b4d485bac74ebe6d948248a : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ed704d0ae1d44bbb8b1b8f8841d897c : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de176b6164364ba4b572db38937eb255 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/983ba540b1d44d61833b58fc34ca4f6a : SUCCESS in 11m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/713f78ef0dc042b2a50596a18902701c : SUCCESS in 8m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d233b021a052491b9916f4d2858f0c12 : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fba5e7f9d4f046f7bb1e1c9ba195555d : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0de564d456654033b6e329f5b7263cb2 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2b9b9f70b284cd9a5cfc31fc61054f7 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/99dcca49a8ea41f39e1b18245cda8bd5 : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/78c03d33c6b540c3a88ca85083884c28 : SUCCESS in 14m 13s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d0d9b852d9ed4defb6470f05e705e8e8 : SUCCESS in 5m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d36ac652acce4d7ea74fce0ad1c80461 : SUCCESS in 29m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/183a9cb5c5264abea46f7d89e4dbe0d7 : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7979ec97754417e9f41cd5ffa0a9e7a : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0edc4398ba0480f91c8dc182c01f4df : SUCCESS in 19m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/33e03009f06e4596b907570ce5b22eb5 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f88ef6bdebb047e8abbf8f446339846e : SUCCESS in 7m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/094ddbe6586f4a9dbcf2487f8d3fb971 : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7554b30bbf00440395500d6d8a170005 : SUCCESS in 58m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da0ddcc4f55a44bd82e4d9a9f019b5b9 : SUCCESS in 1h 26m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0d826a9c65940656e4b0162b1ff6c8bb735ee10a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 18:56:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cfeb523c8510d50e69d2150b27d89c999fc9263a","revisions":{"c5304d540401f0e8fe40febbd81ae6415d9655a6":{"kind":"REWORK","_number":1,"created":"2022-04-20 14:41:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/838735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/838735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/838735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/838735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/838735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/838735/1"}}},"commit":{"parents":[{"commit":"33242c2121a1faae0748b62669a88b8364e76b1c","subject":"update constraint for python-manilaclient to new release 2.6.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33242c2121a1faae0748b62669a88b8364e76b1c"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-15 09:36:55.000000000","tz":180},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-20 14:41:43.000000000","tz":120},"subject":"Bump diskimage-builder version due CVE-2022-24765","message":"Bump diskimage-builder version due CVE-2022-24765\n\nThe new git version released with security fix [1] for all current\nUbuntu Linux releases [2] [3] [4] which breaks current deployments.\nThe diskimage-builder bumped the latest 3.20.3 with the fix.\n\n1. https://github.blog/2022-04-12-git-security-vulnerability-announced\n2. http://changelogs.ubuntu.com/changelogs/pool/main/g/git/git_2.17.1-1ubuntu0.10/changelog\n3. http://changelogs.ubuntu.com/changelogs/pool/main/g/git/git_2.25.1-1ubuntu3.3/changelog\n4. http://changelogs.ubuntu.com/changelogs/pool/main/g/git/git_2.32.0-1ubuntu1.1/changelog\n\nChange-Id: I38b8bbca61f519929ac8625360ea4f10ca75974d\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 53a2f8d4dd986cb5547b5ba48fe3765e654c246d)\n(cherry picked from commit 91561fd4e3776b6f56677e370693760a9dbe3cf5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5304d540401f0e8fe40febbd81ae6415d9655a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c5304d540401f0e8fe40febbd81ae6415d9655a6"}]},"branch":"refs/heads/stable/wallaby"},"cfeb523c8510d50e69d2150b27d89c999fc9263a":{"kind":"REWORK","_number":2,"created":"2022-04-20 15:05:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/838735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/838735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/838735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/838735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/838735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/838735/2"}}},"commit":{"parents":[{"commit":"33242c2121a1faae0748b62669a88b8364e76b1c","subject":"update constraint for python-manilaclient to new release 2.6.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33242c2121a1faae0748b62669a88b8364e76b1c"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-15 09:36:55.000000000","tz":180},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-20 15:03:32.000000000","tz":120},"subject":"Bump diskimage-builder version due CVE-2022-24765","message":"Bump diskimage-builder version due CVE-2022-24765\n\nThe new git version released with security fix [1] for all current\nUbuntu Linux releases [2] [3] [4] which breaks current deployments.\nThe diskimage-builder bumped the latest 3.20.3 with the fix.\n\n1. https://github.blog/2022-04-12-git-security-vulnerability-announced\n2. http://changelogs.ubuntu.com/changelogs/pool/main/g/git/git_2.17.1-1ubuntu0.10/changelog\n3. http://changelogs.ubuntu.com/changelogs/pool/main/g/git/git_2.25.1-1ubuntu3.3/changelog\n4. http://changelogs.ubuntu.com/changelogs/pool/main/g/git/git_2.32.0-1ubuntu1.1/changelog\n\nChange-Id: I38b8bbca61f519929ac8625360ea4f10ca75974d\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 53a2f8d4dd986cb5547b5ba48fe3765e654c246d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfeb523c8510d50e69d2150b27d89c999fc9263a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfeb523c8510d50e69d2150b27d89c999fc9263a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
