)]}'
{"id":"openstack%2Frequirements~838837","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Id0172fe9d2cfe999512eaebedfb4d88917c50503","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Id0172fe9d2cfe999512eaebedfb4d88917c50503","subject":"update constraint for os-net-config to new release 13.3.0","status":"MERGED","created":"2022-04-21 07:53:04.000000000","updated":"2022-04-21 16:33:30.000000000","submitted":"2022-04-21 16:33:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838837-new-release","meta_rev_id":"a63be03bdb2feff80627cf855059ba130a655d6a","_number":838837,"virtual_id_number":838837,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 16:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-21 13:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-04-21 15:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-04-21 08:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2022-04-21 13:45:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-04-21 15:33:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-21 08:19:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-21 08:26:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-21 09:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 13:45:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-21 15:33:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"63326d933f072344f79c975a15f4abfcf4dd46d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-21 07:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d713fdc94ab77cf71d9a39a4d980b08b234466ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 08:19:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8b3c5fb57042437da1d4620a3e8cb2e1 : SUCCESS in 24m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e944fd095406ed4038a9f15cd85170229822fe41","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-21 08:26:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8c12f9602243c319e999b3b006087e6929ebefa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 09:15:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1614be919ac4199a0367d498ed6b967 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fb73d7832e94934bd7f2c60eb7ec255 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/195465f150c04911b20b393aebf297d7 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61f9ba1784344a12b98f499dfb063e02 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7791124de6b945058a6b7369838a128a : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e52dc2d8e929493e87619ae797c8d73c : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/995ea3a84c084c889073d1b824857d55 : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/635439fcdd53462486566b7e197b983d : SUCCESS in 9m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b1909740e9f24bdaa1cd559ee1b9ac19 : SUCCESS in 5m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0260b4f22b704a0194d01e07d249bdb8 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fd9157f975f49abbcaaab8499f66d22 : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a5a69788f5f542e1a7e1ed28ffed2048 : SUCCESS in 7m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/86ab6a5678044e3f975810bb4a423d58 : SUCCESS in 14m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f23b24e282574fcb990c567f02bcfa95 : SUCCESS in 28m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f14f3a69fac4ab9b662c8c2c6c0bcd6 : SUCCESS in 15m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd8dfe9d7d374973ac4007623aea25f8 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90f041e3852f4389a0a6b8eebcedefff : SUCCESS in 20m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c7c3a9f5152b4f968dbdc0e4dc1ed942 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcb80dc7c79245faacdb511dd02ccc2f : SUCCESS in 6m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/00564c74b1ee4c9aad8e9e6356276fe1 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed83e0df8e774969b4af735cd4a273f0 : SUCCESS in 1h 05m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3843b8e92d964f579457e53e90982916 : SUCCESS in 1h 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4c9baf7e9077a19db62906f35d42183d9de7af6a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-21 13:45:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a395ebdc6fdc31ecec4e404ba1f53d0d0c94b14f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 13:46:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9bce5b8de35c460191809158ed5f473c5af2163","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-21 15:33:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47365880ce234ff12b5b556be2cf21885c995cf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:33:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3769f6c6bfcc4841aea51b6c8c38ac79 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7e1d01d37074025ae2abcff055689e6 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc7cec80a8d4f2fb2ceb9885a9396da : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e30b232571a149ac9a6b299acc5541b3 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/084a9fc586554f85a8785fdb380020bc : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a0fe6a7751848c0b47a41e051f7eec4 : SUCCESS in 8m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/efe0ffa223e44fd68c418ce15a2b55d9 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/145a9ef5f0424a72b88374a331a657d1 : SUCCESS in 9m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fee523cf6e094ab18eb6a5ec3133086d : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0154a8fa31a345cfb8f3cbf1547a067c : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f87ac052a70f422eb3f8f29b5b36aecd : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8634e89a48af4ed6af95d6f0a525e77b : SUCCESS in 7m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/10fb838b80ed41d8a85bcb5a5b1ca592 : SUCCESS in 15m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eeabc98ee4c64643b0a0a98d7dc227f9 : SUCCESS in 30m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1eca87b3c9e54da2968a695545461c7b : SUCCESS in 12m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/68cadb479ba447e2a3552e72f29d913c : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40b4de9eaa8141e28227e73da38eba2f : SUCCESS in 21m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/849f542ed6b74c3dac061ca0a6fffc55 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/710d39ff68e34158a93867ab22cbaee0 : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3fb3ba8e9039404a94f89e360ee8485d : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4740827c328e49e58307882082174c82 : SUCCESS in 57m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7521974c82f441b84b6577901b18368 : SUCCESS in 1h 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a63be03bdb2feff80627cf855059ba130a655d6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:33:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29804d3097259fb993c9ffe122ee442524a6db3f","revisions":{"29804d3097259fb993c9ffe122ee442524a6db3f":{"kind":"REWORK","_number":1,"created":"2022-04-21 07:53:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/838837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/838837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/838837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/838837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/838837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/838837/1"}}},"commit":{"parents":[{"commit":"0694e6e38cf53601c05a207353a4c1dc134a4ad2","subject":"Fix requirements-check error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0694e6e38cf53601c05a207353a4c1dc134a4ad2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-21 07:53:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-21 07:53:01.000000000","tz":0},"subject":"update constraint for os-net-config to new release 13.3.0","message":"update constraint for os-net-config to new release 13.3.0\n\nmeta: version: 13.3.0\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I900029863bb6b08141de21cb42d6f9c43fb4e13e\nmeta: release:Code-Review+1: James Slagle \u003cjslagle@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id0172fe9d2cfe999512eaebedfb4d88917c50503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29804d3097259fb993c9ffe122ee442524a6db3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29804d3097259fb993c9ffe122ee442524a6db3f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
