)]}'
{"id":"openstack%2Frequirements~839019","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ie9c4deabaa8b205ae8951d689237af5cf9a6f73b","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ie9c4deabaa8b205ae8951d689237af5cf9a6f73b","subject":"update constraint for ovsdbapp to new release 1.9.2","status":"MERGED","created":"2022-04-22 08:49:24.000000000","updated":"2022-04-22 17:05:17.000000000","submitted":"2022-04-22 17:05:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839019-new-release","meta_rev_id":"dd53dc4d77a41442d44c38b327132a73137d8240","_number":839019,"virtual_id_number":839019,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 17:05:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-22 13:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-04-22 08:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-22 13:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 08:57:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-22 09:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-22 10:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-22 13:55:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c8cce26da8d177107c0e964768dc83a7830d1e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-22 08:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e76626d1306ea4d1abd0089ce3fa173025fca0c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-22 08:57:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1135545b341945b02af883f291d71cfca0d47b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 09:07:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4b1373d36c9245c3a3c3eaac92cb2e0c : FAILURE in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"46abce8cf735336af8b924c5cc42794e99cec7c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:22:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc7cad5d5d045bd8ead99c9100f9ac6 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b451e059c6a49f78eb05480d8380831 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8867602aab4a61875fcbfbb9e46f66 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59fb840540654af79e28422fa86d0ab7 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/982c9b42723446349a8a2a9d57735c9a : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab1be985a40943cb8e793b3086a17ef6 : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/953ab11f0d8442bebfc2bf0ed4e34713 : SUCCESS in 9m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b0c5a91c1c2042c1ae581f09d70a1f38 : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/467a15efd11b4d2cb93f3016c1a7eb0a : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ffae1ab825d74804852ff26e5e70e1fe : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57f9300bb7f94a9ab3c4b66d6536ac0c : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7fa758970e64fb7bf54e0fe66d1a056 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/896a1a75a47e490b9a4aab314a9339b1 : SUCCESS in 8m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/66070c0017fc43d9acdf1b943f314a2a : SUCCESS in 16m 14s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8c9b004528c649f5ac73cca1085bfe90 : SUCCESS in 5m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f5776a7ed4234659b7d7f4a363c6f230 : FAILURE in 4m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0b2c621cf4684ae5a63f3b284b0ddbe1 : SUCCESS in 12m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2657b4f274db421094efc7a0cd417063 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/528ea114df6e4c5a8303c2361b26b5fc : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3c628db7759245488f4c0dfa8968fd1e : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56d43523dd734debaa56f5d7e4a36728 : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/78611e15478d480994a4b3c636c725ad : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cc490db6fb14e2794ea6d4dc9e4378c : SUCCESS in 1h 05m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1360c2cc4024ae8a9044306714eb63b : SUCCESS in 1h 26m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4d1418461bd86a14285c1bffe0cab237e75c9ade","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-22 13:55:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9212d4650ae7e452a4f521999bd78c2bdc1cb061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 14:28:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/21ed246b51fa485e8a23d93aeed7f919 : FAILURE in 14m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3dfb99dac903c4d2fb41b1bf7ad9073bc6be232f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 15:33:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080635ce8ea147abbe5003c320530df7 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59c792ab5cce491296d47e1cafbacd4d : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54b210b53baf4086814a8ba3a7d4057f : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eec4ea7b8ad84a69a8306f5e3ce061f2 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9439c63d979444eba9d3a0731a3d8e59 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/566d422724d145deb402a4ff97b49732 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1622a36701d446859abe89bac08c6023 : SUCCESS in 9m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b0d6b4f645bb4bb8bb5faff3221fb88e : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0b65f3a8f3044706a1cb61658fd8297f : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00c2531173394da8874356243c1e1bf8 : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/40185265a0414e90b17dfd8e86507b3a : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ffd0fbd7f1741cc8ada0eaab479c206 : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3617f9e162c04413825ea9c23ee126e5 : SUCCESS in 8m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4383093b5c43496d8ad64075aef585a1 : SUCCESS in 14m 57s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/aeea9528f264442c9bd59ceb3532b18e : SUCCESS in 5m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1ae6114081244a3696a0c8fa2541aa85 : SUCCESS in 30m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/384b9450143b4f4d8405001a04a5563e : SUCCESS in 17m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cb729c47ec2448fb8c4c645b1c864a27 : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76ed1130734347c1ac7488f8054d055f : SUCCESS in 23m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/914a33e1fd854c5ebfc0efeb38da07b8 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3927333e65c44bd80e933bbf7d92871 : SUCCESS in 7m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db12a2e895f94a87bfb13dd52c0248b2 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8aa1bdef50ab421f9c69301612fb3f44 : SUCCESS in 1h 21m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18dbef88c40c4e489752ad0933334a31 : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6b376589ed3deb767cf069e40bb11484752f01be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 15:34:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"008a9e479cdebeed77d3c089ce4fe19e74d42e41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 17:05:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aad47d37ba243ca8e4832b061c64ec8 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52d8981664c549cc93e9f31b3624f412 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20ae914b487648f09c31ca27fdc3cd62 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60dc6914914647e081a3a92730248e4c : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d6e58bef38f463e80df883daca1f8ff : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b06a68a8933445dd90ab9ee10284d994 : SUCCESS in 9m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bde75ad3cd9e497789577fa687b2c36e : SUCCESS in 8m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e92b39a58f614f349e8881a04253a752 : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8bc10facd0fc4b73af2b74d774932307 : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/294feac17dd74c23a4601f4c5b81eec4 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2ed0f3d06c1462b8eb1c97c97e86cc7 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/12abe9ff898a4a2da2e4d54f1b350554 : SUCCESS in 8m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9942faf0b734273b05d931bb4fe4f92 : SUCCESS in 13m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/76d12c32902a45c28c0ce79e7be4ca67 : SUCCESS in 6m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e29d63e3544f43f69096410c9c7b6b8f : SUCCESS in 28m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8125150f62c54fbd9fb890ec7a0ade85 : SUCCESS in 13m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77ab025a789b481ab0a8da3d6d2a717d : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c904e449d7464fb6a62c62fe02401e4a : SUCCESS in 19m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4548a19288ce4420b840679ac661ec3e : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02a3cff1bae7447892c8fdc97fef6764 : SUCCESS in 6m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bc60197d3d7427797d838d5fe7638b7 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bb62802b224484fac3f97563d95c8ce : SUCCESS in 48m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/539445dae0754a59ad01c6e0d424f9e3 : SUCCESS in 1h 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dd53dc4d77a41442d44c38b327132a73137d8240","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 17:05:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e46413c5ab6fba94ddf219b5efe4d52d0dba95c","revisions":{"7e46413c5ab6fba94ddf219b5efe4d52d0dba95c":{"kind":"REWORK","_number":1,"created":"2022-04-22 08:49:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/839019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/839019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/839019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/839019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/839019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/839019/1"}}},"commit":{"parents":[{"commit":"9f09bce52bcf8944ac035b179a5b90faba0cdba0","subject":"update constraint for neutron to new release 18.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f09bce52bcf8944ac035b179a5b90faba0cdba0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-22 08:49:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-22 08:49:23.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.9.2","message":"update constraint for ovsdbapp to new release 1.9.2\n\nmeta: version: 1.9.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta: release:Commit: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta: release:Change-Id: I07df4e201cb963f50961d49227035582e5e667c2\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ie9c4deabaa8b205ae8951d689237af5cf9a6f73b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e46413c5ab6fba94ddf219b5efe4d52d0dba95c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e46413c5ab6fba94ddf219b5efe4d52d0dba95c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
