)]}'
{"id":"openstack%2Frequirements~839565","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I6a53a5743e9c3e849f489cd4d845237cfbe86b5e","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I6a53a5743e9c3e849f489cd4d845237cfbe86b5e","subject":"update constraint for sushy to new release 3.4.6","status":"MERGED","created":"2022-04-27 15:00:08.000000000","updated":"2022-04-27 17:56:12.000000000","submitted":"2022-04-27 17:56:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839565-new-release","meta_rev_id":"0d8e9e32bcfba8fc49a64a7ca2f1d008d002a172","_number":839565,"virtual_id_number":839565,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 17:56:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-27 15:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-04-27 15:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-27 15:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 15:03:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-27 15:06:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-27 15:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 16:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f58218b3b707b4e73ff0464a6e8386c85004abf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-27 15:00:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb5eca68aca1f7682ecf20cbec8ecbfd929b700d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-27 15:03:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80b4d1dd4cdd8404e8788c94f54c5b9069f00efe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-27 15:06:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b00e45ffd3a678807c926d70a8d57ab97c05f56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:20:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7397b459cf1d48e2bdfc8b322ef9f049 : SUCCESS in 18m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1591cf2ba38c0b355cfa89462ea30484bab0efbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:25:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882092ecd8a64621a667c56200dab574 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9687acefac448fb9aeec23dd9293d19 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e08662530d4c422cbbc83e390e17f21d : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fb8023ed1734d2e832c1484386e1213 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ca483e28f1d4f339d3d81fded2728ea : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c61071856bab4b94956635f2dce88514 : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e91a7160e9f4734979b2743739fc40f : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1fb56d4164e9436d81d6a9b1932494e5 : SUCCESS in 9m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4dc8410b017c45d8b80f72139e53d4a5 : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ca19f2cfaa1440288046f2f910772687 : SUCCESS in 13m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1996f3212c94bdea62d0a9316d879be : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e2da6b516782412f8a8d14ad5b1bc55a : SUCCESS in 8m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e1a8b48b4b1c4625a5a181f2f6e77a26 : SUCCESS in 17m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/df6fba61c2524784994ea9726ba1cdcf : SUCCESS in 30m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/56c5499711a242bf897c9a60d974e1fd : SUCCESS in 15m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5b7abc8798a47e9aee2ad95eef482ef : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0eba4fb379648bc876d6e6e9d5cbb80 : SUCCESS in 15m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a8715fa4f1aa4f95b26037d4d9c6cc90 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d521a2d696d445e78096fe714ea555a3 : SUCCESS in 6m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/455494bbab124fecaf9a355fc3bcf2cb : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8706c8b8ee644fa8d7214030eaba7eb : SUCCESS in 50m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/910bc3a90a5c430e973d88807e7d6ae8 : SUCCESS in 1h 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5f6b9ccc28f2971f46d3bfeb1cc1701ed87ed428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:27:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df1e126344b8b9b108b2dbfba09a2c0b04c9816d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:56:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e9c9f33d2140afb332949443ebf98a : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff734d402902497badfe4c0b9450e309 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ed38b6041ed46ad88626804e3306910 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d54a284991174613bcd95a210367b86e : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb1ddd350c024c9985106895f4bfab14 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce36022da65a4b738a70c17b572235fa : SUCCESS in 8m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef2986ba03544ad39e36d369d7701d35 : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/197cdd84015f4cf18226cf4abc839d6b : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6db294e247794ecab2032e098acd6e43 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/50a448d3ed254df182c81732457c8ada : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d7f56fb6ba54008a5d4a2ac18452a31 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d17f25ffcb9b493393728f17341de59a : SUCCESS in 7m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a5420f8b018e464f96c67f5925c4c767 : SUCCESS in 21m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3206ce8d4bb348f7b5fd7eb2b7919f23 : SUCCESS in 20m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a0e95c5dc2ba4dc3845b2aace0810056 : SUCCESS in 13m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6694d563f2a54dfea4e7ac1bab0fa762 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5ee7dff13bc4f85877065475969e1f1 : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e14850d420bf46f98a16159ba2ef1bf3 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cc72c5fbe12462d8071eadec4c5c7ac : SUCCESS in 7m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b7f6841cfa7b47faabf9a50dd6099d3f : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fba6136a54448fa80a19b5ee0f4876d : SUCCESS in 57m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b0ae30a170443af9b803ec1da481b1c : SUCCESS in 1h 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0d8e9e32bcfba8fc49a64a7ca2f1d008d002a172","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:56:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c7b195032de8e5d6618e96a9d7e3cbb3a9a16a2","revisions":{"2c7b195032de8e5d6618e96a9d7e3cbb3a9a16a2":{"kind":"REWORK","_number":1,"created":"2022-04-27 15:00:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/839565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/839565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/839565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/839565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/839565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/839565/1"}}},"commit":{"parents":[{"commit":"94ade82348cd30511728ad2132881a8c0c930016","subject":"Merge \"update constraint for python-manilaclient to new release 2.3.2\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94ade82348cd30511728ad2132881a8c0c930016"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:00:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:00:07.000000000","tz":0},"subject":"update constraint for sushy to new release 3.4.6","message":"update constraint for sushy to new release 3.4.6\n\nmeta: version: 3.4.6\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\nmeta: release:Change-Id: Ia5662fb9f336cdf52bef2e42df677f20043343ed\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I6a53a5743e9c3e849f489cd4d845237cfbe86b5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c7b195032de8e5d6618e96a9d7e3cbb3a9a16a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c7b195032de8e5d6618e96a9d7e3cbb3a9a16a2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
