)]}'
{"id":"openstack%2Frequirements~839577","triplet_id":"openstack%2Frequirements~master~I20ff264af4288e6cd7d5b9aa231764383f2a7605","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-05-05 13:21:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I20ff264af4288e6cd7d5b9aa231764383f2a7605","subject":"[DNM] update constraint for oslo.policy to new release 3.12.0","status":"ABANDONED","created":"2022-04-27 15:35:46.000000000","updated":"2022-05-05 13:21:58.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"975cc8c792487c5e284773efb757dc0074a83797","_number":839577,"virtual_id_number":839577,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-29 07:05:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"disliked":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2022-04-29 04:57:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-2,"date":"2022-04-29 09:28:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2022-04-29 05:08:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2022-04-29 09:28:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 15:54:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 16:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 17:17:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-29 04:46:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-04-29 09:28:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"1a6b05e10ec95e3c9c8455972993dce4cb858a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-27 15:35:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cd6bd120b393827b81e093b08ccf62ef2a8cd0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:54:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/91babf9606b4480a960ea89aeeb976f2 : FAILURE in 16m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2afa27d8fca022d557b752981283cdf35d303ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:56:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c75ce642c54518a10a14270259101a : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5719e4a3452744df81021c52ee3eca00 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ad4a3072b1b4e87a2cbc0cb2042b78d : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63bbc982b826428882259d7e8a755c21 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61684ce6d2244d008e24273f4145626c : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cff646708fea44aba8e7cb1b34af25c8 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/417515f6cba947aa98f5be96e1a85534 : SUCCESS in 4m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e5aa124904fe4e3da1716578edd13665 : SUCCESS in 10m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e14940c5f1e24781806797a7562824ee : SUCCESS in 13m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dc657063cd0140e792871f81959f6034 : SUCCESS in 6m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/27b3b4504e564501852a909fa56a9ff4 : SUCCESS in 10m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/63d5da4713cc41689c55fa28ab1e24a0 : SUCCESS in 5m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c71d7b54016d4b71a0fe4b96069515b7 : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bb749af01a324ff9a0d507e504023c2b : SUCCESS in 9m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/39fa692f29d14b348c84fba6db979f63 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf94b6af90d644c99845967d835c496b : SUCCESS in 9m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d99f33648cb948bf87accb60a1a7041a : SUCCESS in 7m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/491ab564028a4ba8a96f5d2391fac0dd : SUCCESS in 14m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2a1f344fa3174f7781f1eca29634b647 : SUCCESS in 5m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/da11382e04364350aea20b542786bc5e : SUCCESS in 10m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/11aa853461674bbdba0c1ca2f4d7ec2e : SUCCESS in 5m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eb5cd85dce5747e6983f6f9e773bd560 : SUCCESS in 22m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a6669122c1da4df1ad9576ce3668656d : SUCCESS in 13m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3744c822c253439dad2869378493776f : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa80cb45b4a04e2c8da1a27b5852647a : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38a14d2a9b554f71a47ee1f7158260e4 : SUCCESS in 18m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e4edf038961c471cba682ed60d017287 : SUCCESS in 6m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e5de730d475c4fbabd93262179dc21bd : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/138b2ea51ffc4c35bdffcb2f515359f6 : SUCCESS in 7m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9a10953c4df24df988d90c13a9c1be3f : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06a803588b0f4c5c96e958769c7b479a : SUCCESS in 1h 02m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e536a524a3f4df2b0722887fd8d8e92 : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"95fe484c2be0491792277e87edbebc26015b57f7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-27 17:17:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"907537f85e5f92a0731492425467550255edd0db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:19:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b899f82f253e321718938ef5908347c8e122bc79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 19:16:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe437334bb23402aabb85bfc26792dac : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a0c8a3a5d2943388136908c90fc69c1 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e55a880ab49c4b989b8b4b68b27fce52 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2703aa1da1624d1da58811b87ca87a6c : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10793823d9bb43cfb246a708587c9dc5 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9b9dd329d7334a4a9dcd693d4d504636 : SUCCESS in 5m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3831b7ec590f4219a5382d190ff73baf : SUCCESS in 11m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d370db09325c4e1ab62ba7291860c183 : SUCCESS in 15m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a67a250e0b874772b41f5453c03008a3 : SUCCESS in 11m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/594141b392b8422fa70a2a5e2cabb2e7 : SUCCESS in 6m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f8f0604961994156a6ee5bff36255af1 : SUCCESS in 6m 36s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a0367a50c1754a98a26171e5baa96014 : SUCCESS in 10m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ef1d8a38ed3d47a3b9bd4a0e43e12984 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db2ad150bdff4226bfef681ae8059344 : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8d6ac89ad16b4c93b166a1eaf0bd0fb7 : SUCCESS in 10m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/be5911ef8c9e4f6b9f86a49d8d98f450 : SUCCESS in 12m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ceca77a05e48480ca109d66a882e5c3a : SUCCESS in 6m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8573354274184745a6636ff7bc13f4cf : SUCCESS in 10m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7c1d6e41a9384fc485c7c1d2c40b28d6 : SUCCESS in 6m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e17ee68f0b044b9295b56a46a73603da : SUCCESS in 26m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5f662f015e4a4e15b83708b824681bba : SUCCESS in 16m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6e62093d76334ae68e2a4656d9afa5b1 : SUCCESS in 7m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c422aa2dd284a84af237bf83d3d7846 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f14553b16b264745b5b447cc65fa7f76 : SUCCESS in 20m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/449dcf7252f640139c6d34c89436db76 : SUCCESS in 8m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bd23ef24aa2a406e87d5a701545bb2b7 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71c1be47939549d995fb18a372ca3845 : SUCCESS in 8m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9b9fa21eb53a405eab21447abe6ea5ab : TIMED_OUT in 41m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eabc3cc520e145ee91a7641460caaa59 : SUCCESS in 48m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cedc5e994bb64b7981a45bbd25de0eb5 : SUCCESS in 1h 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"35440ac38c8d5b66a238ba7e7b4f80dd29be8bca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-27 20:26:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"30fa03e43b7576e7f54c3038ce54d6b5440a465b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 20:43:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ae94734e497b4bd8bad98b6b92c177fd : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"790c8d3ce7e27b7345f6dc0904b1ec6e516e93ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 21:56:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c08f59018b04f66aa68ec1afca52543 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd0fc774c21c402cb5918be244b3ab1c : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12a00056c84640cfa5341fc7f0c1e7fa : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47fc54e98cf84b389fc0c93058078617 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe13911ecef5441fa8fb9860e452daf7 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90535c78926340e4b39a1ed14545c090 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3cd92a15d59c4735b60970101b45546b : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10b670a6bbce456eb94722b72b230ffe : SUCCESS in 9m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/be26836d47894c6b8368bdb048e331e3 : SUCCESS in 13m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dcb4c4820d664b4f9b9c409b70f5b0dc : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1aa02b52ef2e4341a2d36fe59e9df1ad : SUCCESS in 8m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ec0c9b730d7444b7a8747e981ae7e332 : SUCCESS in 5m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/146fa09e0a7f45178cca457cc5212b90 : SUCCESS in 5m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b112984d84da45cf8f3d6c87d6480fb2 : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9462ed60af2341edbe3941170e0b52dd : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1aa2094fc7dc42f290bddb4d0681e9f2 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6855a07c14dd4217824c003f6f3f932d : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a22adbff6821482eb671035739dab423 : SUCCESS in 9m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ec894ed0d1df4ab58c53bb854eecdb2e : SUCCESS in 4m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7a27ed4b6f564e08968d8b57526d95d5 : SUCCESS in 9m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2d01e0d3edf5444f955e6046889a2af7 : SUCCESS in 4m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1e7d3168595645f2b8e68f72bca043e4 : SUCCESS in 23m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6435af0faf704e32974b4e270f2135fa : SUCCESS in 12m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a118ab3eeddb44cc8af4c3102b0eb2bd : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60387ffcb5d942c98e208009b764a786 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24b22542a4594d208c26fe3c902f592b : SUCCESS in 19m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b8d03c4a9e444e4ba5ac96d3c187b663 : SUCCESS in 5m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/da1a1a5875e14b208cd2451e69146162 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b3d912199f94d7282daabb4e2578031 : SUCCESS in 6m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1964d3d6d9da4af1be492cd639f83082 : TIMED_OUT in 40m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/300846c3cb394d53910aa7df59c49226 : SUCCESS in 43m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eab0077f16a345a29d27236eff63ae6a : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"198869179bdb862e5d5b560382797299d4bc5dcd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-28 18:31:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ba0c248b975b604e0513706006189d3d149ba15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:08:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e9c99844f59e429384a9d72477e78360 : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"27e2beeb5eecf680a14ce573b70f9c43a95ed0af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:04:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b39dcf2ca5435e9fc1f118fe3f771b : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6778da219b64ec9b630be05bbed8887 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad099fb0acaf431aae38cb645a0c9049 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b192308d41474be49b429dc79bed1d81 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82358e2ead3f41258ec9f7d5088b675e : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e99d0e2f1de945abb0e78865ca271b72 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/267b9a8fa03542ed9258dfc14487f1f4 : SUCCESS in 4m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1d1d9e712774d79aee2f7a99151c4a2 : SUCCESS in 11m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fd8843531ac140d6b788e1c222a2ae3b : SUCCESS in 12m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e1822b4e42054a3f95a8a19cd1de2960 : SUCCESS in 6m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/21d88623845c4c2589d046dd96b5d1bd : SUCCESS in 11m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/17bf46705e234988a7ea44f89450746b : SUCCESS in 5m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/51fc091fd40b4a6d812abfef9f0de641 : SUCCESS in 7m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c94a85a41a014765b721399e718e354b : SUCCESS in 7m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/10ea21ee54dc4ed5a8f606f474b1ddbf : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72bbfa7fa7cc4c3f9d929258d3a19185 : SUCCESS in 10m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4bf4a0cb1fc47a396c6b7d3b675508c : SUCCESS in 9m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/78379a04bd1442868cbc224edc57c11f : SUCCESS in 13m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9780ef0ac8614131b01526113758e04f : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a8fc0e1aa82f4ef2a26f5e0556410087 : SUCCESS in 8m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/61ee620445004a8990f1ab2ca87179d0 : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a023ffddcb2941b5bdc90974457c8d0e : SUCCESS in 20m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0c98d9b166394c488b5d49d7c21f300c : SUCCESS in 10m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/10ed137a2a8f4611b9dc5907ae15c6c1 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f2a8cd50281446d98aca31caef899857 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45d58c3cf28f41fabae261b4b3fffea6 : SUCCESS in 16m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/62399b77429d4d739016d26d489abe92 : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a9104ecf60fc4d66832d72168f123033 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62f24bf554464b0c931eae452b853ced : SUCCESS in 8m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/56fcb8b97e9a4dfa8cd79e9205fb813b : SUCCESS in 14m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1b7dd13405843159254c948c1f61f76 : SUCCESS in 1h 00m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f831c0685994042b359f6b1a186146a : SUCCESS in 1h 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6b64717fd487a8b2de0f9dbc7138aa5846beb136","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:04:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c3c79afa6dead9f802edbac6e15815dd57aecf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:42:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c50705d16464383a715dee6e433d589 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c49ef847264410b740d52b21d0cd5a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6729cf42e7445590b78e44c65ecad6 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/021e598f6d34471dad68b9677c97abd5 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac0e541695224af798c744da2ed0c568 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f2d61d84bd64094bd4f1e72feee51c8 : SUCCESS in 4m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6123f0208e544169a77757356dab01c4 : SUCCESS in 10m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0c4df14fb7ef4f2aa0ec13ec6412949f : SUCCESS in 13m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/78b3d47e2ff84c96b63b11393b2a53a5 : SUCCESS in 11m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0cdde26fc3704814b42039ef793228ca : SUCCESS in 5m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/65eef8b5d1ec4cc896d50695f28c9806 : SUCCESS in 6m 13s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/64a930bd32834d4db9e3d06df20f3770 : SUCCESS in 9m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d9d54aa771c34ad9bd7dd477cdcbe519 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/30b62f67d11f4929896d455b46b5a6c8 : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/103c3a7f09f54480bebd8d4e6408d9c3 : SUCCESS in 9m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b3e0ddcf8fa24297bf2462c2f9a45a69 : SUCCESS in 13m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b098ec0f00a640fb812e198dacae3ba0 : SUCCESS in 5m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6dd410a8be4645dea04233086d2049bb : SUCCESS in 9m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ce91f2578dfc492987f7ba56fbc4a7a1 : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9d7c32aacf1e4f03a49f12b34762665a : SUCCESS in 27m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b5e2cdae1eed45cd9295e0944a270d93 : SUCCESS in 12m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/00cc4cd510194d5baa91245a784ea766 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0115df7b249742489c111115a012ab79 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/477b27c01010493894c20fe5e364be70 : SUCCESS in 21m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f4784a6636414ae3a4589e9f6918da1e : SUCCESS in 6m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f269e338275b463b8a469450dee0efb1 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e4771fb8ac54cd58287c733f1b47984 : SUCCESS in 8m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ad0153f1aeba447ebc4c349cb9d5d566 : TIMED_OUT in 41m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eaf6edabb644dd393a64d12ef64caee : SUCCESS in 52m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/871cf793a77b45e1a15780ca7b42d9bf : SUCCESS in 1h 20m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8fe102c080ca327fa8e998df062f0663a9f23c40","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-29 02:56:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a19bb0ed21fdd6989a077d05008db161d137c141","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 03:37:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e35093e0e07b4167aa82375c44877b1e : FAILURE in 16m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8f373a704045eccc4196c498f986b49a65f85365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 04:41:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2065046270470fa8b3fdbd033ef7d5 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122b4d93a3604b34a0cfef25cecec4c3 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd9438d371b848529a86b7fd83015960 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d1cc3dae29f47b299f9e78c444e2d29 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/802b1f409c0849e1994b3ba69376ae66 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eea8b19bcb704537a7c172540ab0497b : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04ae4544290a412887e20db8c4f45621 : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1b73810d72444c6681e41c85666c2773 : SUCCESS in 11m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8f93faaebba3482c9886d7515c940746 : SUCCESS in 16m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2762e6a6125647df83890bc692b20f26 : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/41afa011e0984c52a915ccf461dc6823 : SUCCESS in 16m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7679ce0fa7c8476e9dee96fde5f250a5 : SUCCESS in 5m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3189ad2421314a2da0f21e7b1a4acc50 : SUCCESS in 5m 35s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3fbf752e3c354738b8d49b0e07d08482 : SUCCESS in 9m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0cefe04a64cd4d798bd10f3d3ed04baf : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c14bdd4482bc4bdaa42459a058e68bd1 : SUCCESS in 10m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ed9e477383ea41619711c7a6096f956a : SUCCESS in 9m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8a66af26e8ed4d50970e60fc73a660f5 : SUCCESS in 13m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d2a59834326540cabe407a3f858241c6 : SUCCESS in 5m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f73843586421426285ad0f2b48a9fcf9 : SUCCESS in 8m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1fe98bf49ca44938a258feee0cad120a : SUCCESS in 5m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8af8e60b93eb47a88add9b91e2608a14 : SUCCESS in 21m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e09056008e9445a49ccf4949376cd54e : SUCCESS in 12m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/88b849347096426d95725113c869fce7 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a039a09a4e894cbc826c671b1d44be6b : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4486c2803c4d44b4b163035f305c15de : SUCCESS in 20m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7effaf6a07ac4ef19be1835c7b8b306b : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bbf526a32b024d4a96e58dfa8b9503bf : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8676090067424c3f8c7c8c707282b057 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/590b5fa8fc5d42f0bd13c85982c88c51 : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a19b40af6fd4fe6975f388e93ae4c68 : SUCCESS in 47m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bccd4cdd963949b29ce648ea555ab47e : SUCCESS in 1h 30m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2e50cfcc73f839416e4c3d6759b01ad682072914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 04:42:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8ef5af237794e4247ec70287fd4ee9435b8987f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 04:46:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee47d0f0880cda2497273de47e9faa51796979be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 04:57:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fbd473948f868fe254c9b8b33f0437c7ec074d1d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 04:57:20.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"48ebf96e851dcdbdcf211492240553fd9c69ca61","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-29 05:08:41.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9d608638c416eb36402004d32ab9d1b26df34e0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 06:18:47.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/259d92e604bb4ddb8754300dc9069359 : FAILURE in 13m 59s\n- build-wheel-cache-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3bf01a0a87f34e6ba68341cdc26da31f : SUCCESS in 48m 12s\n- build-wheel-cache-debian-buster-arm64 https://zuul.opendev.org/t/openstack/build/777276b320dd41d4a1f3fd5dc4e2f459 : SUCCESS in 42m 38s\n- build-wheel-cache-ubuntu-bionic-arm64 https://zuul.opendev.org/t/openstack/build/db11fc2332db436786213c76aa79adca : SUCCESS in 45m 23s\n- build-wheel-cache-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/139a9d45b826427ca2f35c13dce21caa : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":2},{"id":"448c52ff04f51ee4cb17efd09e280a19d0f6b3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 07:05:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e362d496d4504f41838ad47e3a576b93 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c45a02075ac486e878f57edc2f83fe5 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab0522a3fad42f5879fe1aff4c8adf3 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eef0a333a75341d994ac8405248e3d51 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bd86b7c196e410285bc655132ee0e86 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abd066f908804dc285a7142f61827980 : SUCCESS in 5m 38s (non-voting)\n- build-wheel-cache-centos-7 https://zuul.opendev.org/t/openstack/build/748d4f74a10c401c8a213fe64f204407 : SUCCESS in 1h 42m 01s\n- build-wheel-cache-centos-8-stream https://zuul.opendev.org/t/openstack/build/a668b39a621643f19e4a93a074ec873b : SUCCESS in 1h 00m 23s\n- build-wheel-cache-debian-buster https://zuul.opendev.org/t/openstack/build/3c336b57bf704236aa42585276ed5629 : SUCCESS in 1h 17m 07s\n- build-wheel-cache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0453e8bbd5304f358d4014d4cd15a0ca : SUCCESS in 1h 01m 13s\n- build-wheel-cache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48c9b2b5812441d79ec26d2d9ebde20e : SUCCESS in 57m 51s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c7d040decf14bfca8557a92d6d05588 : SUCCESS in 1h 22m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6395e5e80d7441139d57dfbb6a7a9b0b : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d90a465f07304b5d97db3349744fbf93 : SUCCESS in 3m 54s\n- requirements-tox-bindep https://zuul.opendev.org/t/openstack/build/a7360a62febd4cba8df6396b5a7b2e09 : SUCCESS in 3m 14s (non-voting)\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6746601fd57a42b791988f93de4a9859 : SUCCESS in 11m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7432a03770bc40c7bf49e3d3bbb8c8a1 : SUCCESS in 11m 58s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/7275a417a5274cd8b105a6f7277645af : FAILURE in 4m 08s (non-voting)\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5b9584a3fc3649a49c68954517e69351 : SUCCESS in 5m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6e34674c44dc4df8873268ae70a3d747 : SUCCESS in 11m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/34e48f3c76764113a954430e77e70af1 : SUCCESS in 5m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dc89e0d5ebb0404880860282fcb0c1fd : SUCCESS in 6m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6e2864eb108e4007926ec767cd60399c : SUCCESS in 8m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/50e412f3a53b4106b4bc93401f0c55b6 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96e02c71f5ea408ebb24af00146e8677 : SUCCESS in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f4e7dd29f6174cb6a3185c64eaddc679 : SUCCESS in 8m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1c73291d071b40e8813c1411ddcc6e6f : SUCCESS in 12m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/182c4e2b980449de9b9d5fcaa5b7fc1d : SUCCESS in 4m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7f2e847e4ebc4f4f8cfdbaa0fb3ce69b : SUCCESS in 10m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b8e835405f064edd8544289fb92f2a98 : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/51da2383e2824f19a92abf184c4f19c3 : SUCCESS in 20m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d8c21b174dc74672b1734a8b9bc31c32 : SUCCESS in 13m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/41321cf5f54c437c8e8c366fdcbee9b5 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18e96b4468974b6a9728f4bd5773b217 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f224ca5c9faa4ec2a63b8f90976193ad : SUCCESS in 18m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ef17b498d16c46ff93ab3f195cc430c1 : SUCCESS in 5m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b43bd59fcdc04a6e8f4eb4c247a0946e : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/617843883b8e4d9ca136aa19784f2a75 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0accbc9da3574b63a273a21314407693 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12517cd70b3649809afb333df8a57c3b : SUCCESS in 50m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7241fd654ca242d6be2e808a19355e7c : SUCCESS in 1h 42m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2f0f3af39f1908c6c1b2a5f4f0f4cdced49bb26f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-29 09:28:28.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d568bf5305fb3b9628809d40c166d5f64130f73","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-29 09:28:57.000000000","message":"Patch Set 2: Code-Review-2","accounts_in_message":[],"_revision_number":2},{"id":"975cc8c792487c5e284773efb757dc0074a83797","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 13:21:58.000000000","message":"Abandoned\n\nnot anymore needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae21dfbfd839344dd828a37903acd7ca480305aa","revisions":{"94e853adfed68dc54c513533d404ab488d03afdd":{"kind":"REWORK","_number":1,"created":"2022-04-27 15:35:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/839577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/839577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/839577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/839577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/839577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/839577/1"}}},"commit":{"parents":[{"commit":"ba7326aa20d1b76fc49f27447b51e01f8cf8aab9","subject":"Merge \"Add Python3 zed unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba7326aa20d1b76fc49f27447b51e01f8cf8aab9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:35:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:35:45.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.12.0","message":"update constraint for oslo.policy to new release 3.12.0\n\nmeta: version: 3.12.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I7add61a26d302f1247691582fbea10bfb1034717\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I20ff264af4288e6cd7d5b9aa231764383f2a7605\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e853adfed68dc54c513533d404ab488d03afdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e853adfed68dc54c513533d404ab488d03afdd"}]},"branch":"refs/heads/master"},"ae21dfbfd839344dd828a37903acd7ca480305aa":{"kind":"REWORK","_number":2,"created":"2022-04-29 04:57:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/77/839577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/839577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/839577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/839577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/839577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/839577/2"}}},"commit":{"parents":[{"commit":"ba7326aa20d1b76fc49f27447b51e01f8cf8aab9","subject":"Merge \"Add Python3 zed unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba7326aa20d1b76fc49f27447b51e01f8cf8aab9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:35:45.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-04-29 04:57:03.000000000","tz":0},"subject":"[DNM] update constraint for oslo.policy to new release 3.12.0","message":"[DNM] update constraint for oslo.policy to new release 3.12.0\n\noslo.policy 3.12.0 broken the docs job. 3.12.1 is being released with the\nfix.\n\nmeta: version: 3.12.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I7add61a26d302f1247691582fbea10bfb1034717\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I20ff264af4288e6cd7d5b9aa231764383f2a7605\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae21dfbfd839344dd828a37903acd7ca480305aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae21dfbfd839344dd828a37903acd7ca480305aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
