)]}'
{"id":"openstack%2Frequirements~839581","triplet_id":"openstack%2Frequirements~master~I2c8248b09dbacd72ad619ed088895a988d9cb8eb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2c8248b09dbacd72ad619ed088895a988d9cb8eb","subject":"update constraint for oslo.db to new release 11.3.0","status":"MERGED","created":"2022-04-27 15:37:14.000000000","updated":"2022-04-29 00:25:49.000000000","submitted":"2022-04-29 00:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839581-new-release","meta_rev_id":"e6f19ce74485e7a8f1d6cd8c100ee09c7b525c87","_number":839581,"virtual_id_number":839581,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 00:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-28 18:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-28 18:31:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 15:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 17:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 18:31:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0b63807cfb34b3fc58c3dc520c88786482edd096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-27 15:37:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"710bba01d92ffc4b79cdb35c60ba021615fb1efd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:55:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/508a34f9a9df4263a813dc1950eda4f2 : FAILURE in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cf0e9576ec1955f0ea4c0e6cb1c947b961486110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:15:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dad11aa046147c2a863d7807798ef7f : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7dc4945206246e898b0dda0a4ca6e1b : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1997ef906c604b1f913126fabe64c0ac : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dbf77e500cb4f1fbbfaab618ab4e42b : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fbef87204834f028a1395c0cd8ae68a : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d93b171c75db4a3e82cc7b3c83f3f084 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f147ca1fb63f4ef38175dc1fe8161e31 : SUCCESS in 4m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8230e4f0fa21453f86342409f63c209a : SUCCESS in 11m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0b9391cb3c4649c9958466c60440f373 : SUCCESS in 14m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/edfe829e5f87487bbb5d5ed94411ed19 : SUCCESS in 5m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0c56295c94b548c98633d3a5f98a1d2d : SUCCESS in 12m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9535944888444278a06dcdb66965a363 : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0fbb8e1424d046d09fe38a620cb0a5ec : SUCCESS in 6m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6598710f5e3449798f85ef19edc82d5a : SUCCESS in 8m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/71d54fbef48d4e9fb46933bd9745e521 : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9ba2cbb717a43f6822ff2ec41957442 : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/80c9c016d5ff4ec2980a57d37d6802bf : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7df3c94ab0154ef9bc3b24821a1306ff : SUCCESS in 11m 09s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a118a26b002f495aa4653ddc8bb1d8b4 : SUCCESS in 5m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f236132964b34bcda147e903619c64e9 : SUCCESS in 10m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6b8d8f5ba5e44c449771b24c4664a9ea : SUCCESS in 6m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/235fa8d986de4f699be4a7846ef4c5dd : SUCCESS in 24m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8011c6e247e04aedbb19a48c603b4ba3 : SUCCESS in 12m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/12cb37f66f844e8ab2438949e628197d : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eedd9cac06f2463a86d0382e5489aa87 : SUCCESS in 5m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fba04957b5744cb4830b10c8bef1b41d : SUCCESS in 21m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f04f5384e16b47bcbd89fb96292d436c : SUCCESS in 7m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/56163c874a5c43fe947e2dce62a3a00d : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/777b5227f00d48c7b5c394a986826cce : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3e982ef224f24198bef8364169195678 : TIMED_OUT in 41m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db4876df1ee54456a0515d3cf2448495 : SUCCESS in 49m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/febb6e9d3bed4e97b6f2354a065544c9 : SUCCESS in 1h 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f9bde4ef01e9012c5fa9526cc90ba14f9a470fc1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-28 18:31:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c46acb2773baf7c3225c6d37f3a581661e0a3c2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:14:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/be60beadde1849f59d96e66dfe948b70 : FAILURE in 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2e0f94ff1ffeacb934e5de5782dfefde4cdc7794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:42:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120d390c7f314e3dae4d09941cfd1531 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad477d3ebaa7493a9ff3b24893e4b63f : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48866838493142aa814ed661c8f99c9c : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/116299e690724947bce884844b7588f9 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa34813c6c9a4e5bb1cf510302512446 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10aa958063a94caf91cd2e2cc3219a47 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6e8d4f332f54fe3983cd252721936b7 : SUCCESS in 4m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d14c927595f4349b1560c781004ccb6 : SUCCESS in 10m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/de0a00f8114a4ec29717c3ed6c6d4b31 : SUCCESS in 16m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1ac91d94f98544638f77759a8953e780 : SUCCESS in 6m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/12a28dc938ad4f998884d069ba54933b : SUCCESS in 11m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/62ad42ce565c464682456794caaaf08a : SUCCESS in 5m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c5795c096c664acbb87c0a8a871ef2df : SUCCESS in 6m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/72ccaf2e458d4b59830a2f89929dc633 : SUCCESS in 8m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/11b75f9634d7414d918fc309708b6b1e : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9851da4c9cde467b90cfdcfd97919ebe : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/080f3e5b2425469489432e3a08483db7 : SUCCESS in 9m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/67b9c45863d344bcadc7002ef0b62695 : SUCCESS in 11m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3846f1035659491aac16813d90be2939 : SUCCESS in 4m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2052b8cb16e94c09af43663a12cfe701 : SUCCESS in 10m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/08ec47beb30040b4b56c492f15bb53c7 : SUCCESS in 4m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/88f73b89ec1344589d4cb1ea3090add4 : SUCCESS in 28m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ff8906edceb44b04a5b244497f566d47 : SUCCESS in 11m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f007f483d2fc46f7a87c95e69ed7c890 : SUCCESS in 7m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24c1b1265e7842bc9b392577bbd925e7 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e44359af2444a6dbf48ddd47c27e5e6 : SUCCESS in 21m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6434da9966f14ac29cd6165b0e5b3dc7 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/abffcfb94e69460885eb55d5886b2362 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcb153079f184c5ead91996587937415 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2b5ac9f8731e4614b603625c199c0bc2 : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/201506694d9e49fc84a99ee645f01d78 : SUCCESS in 35m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dcc80c613884bb2ad0266b0dff451b7 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"06b7d3c679377e71a0e32fcbe693ce517a63201c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:43:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d669e461dd35547fa57207b48bbbeaca4bbc0e4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:24:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdadf15af17d42fea859cc2c77482c74 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9304379932ff4c07bb943e110239ec4a : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecfb29a6da5d4f2790a5ba2b17ef13e6 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4848ff2437cb4d7f81dada8957beb8ce : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d698ba57d1e24b88a0b6e8c384580d6f : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff2c072a7d1d431eae86db2b09a00ff1 : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d76c0dc8de054a74b54a090b4099840b : SUCCESS in 10m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/775f2f90386245e79befaf43e4189950 : SUCCESS in 13m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b9dcc4ce56eb4eeca4f91fb40856fdde : SUCCESS in 10m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b9f3b32be06549b28198017dc64fe728 : SUCCESS in 5m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/86d561b7e5224551bdac6286a5516c68 : SUCCESS in 6m 22s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5ac1c789465449bda36fd9c9f039c453 : SUCCESS in 8m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0314221d887e426c8cc93b84b640208f : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/903c29ab5ddd48b4bb4d1c5281c80bc5 : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/075741b8a7f5419d8c280fc24db48602 : SUCCESS in 8m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e20ec781103e4125841d123959058700 : SUCCESS in 12m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/80e8a2110fdd4386a327e9ec9a4fc5d3 : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7928d42af5c54e5997ca6ac4ba410be4 : SUCCESS in 10m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b0842e94d85044f8911eb4fe11fa7f5e : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dcaa0325328f4d42a79cdab68f3298ea : SUCCESS in 21m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b763e8bc0f5d4d25948404de9ce045c3 : SUCCESS in 12m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/35520835d16649f781e416c4d6c43308 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/115e7f360dfe4082b58728b6de624c05 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be816108f74b46c9beb1896ae168053a : SUCCESS in 18m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f4382c8cf4b04f268b39f97292fbd15a : SUCCESS in 6m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/70a965daa6054d4c986ed4ef3f2f92d1 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d7db9fc3b644fb4b723112de8905472 : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d1e1f9b5e2834976b461ef92239158bb : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce65a11cfcb843b3bc5d89b995375cc5 : SUCCESS in 47m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94f63d3a584f4834bc65dc716a635a0d : SUCCESS in 1h 47m 21s","accounts_in_message":[],"_revision_number":1},{"id":"10db365840a5309650ba2946888005773c34bd2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:24:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6f19ce74485e7a8f1d6cd8c100ee09c7b525c87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:25:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d723ce77c24500b90e266248544393 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca3128eaea5563f87164d7921ae248548e9f195f","revisions":{"ca3128eaea5563f87164d7921ae248548e9f195f":{"kind":"REWORK","_number":1,"created":"2022-04-27 15:37:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/839581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/839581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/839581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/839581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/839581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/839581/1"}}},"commit":{"parents":[{"commit":"ba7326aa20d1b76fc49f27447b51e01f8cf8aab9","subject":"Merge \"Add Python3 zed unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba7326aa20d1b76fc49f27447b51e01f8cf8aab9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:37:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:37:13.000000000","tz":0},"subject":"update constraint for oslo.db to new release 11.3.0","message":"update constraint for oslo.db to new release 11.3.0\n\nmeta: version: 11.3.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I7add61a26d302f1247691582fbea10bfb1034717\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2c8248b09dbacd72ad619ed088895a988d9cb8eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca3128eaea5563f87164d7921ae248548e9f195f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca3128eaea5563f87164d7921ae248548e9f195f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
