)]}'
{"id":"openstack%2Frequirements~839586","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I84b8e11c602b002896c01812381772f815486ace","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I84b8e11c602b002896c01812381772f815486ace","subject":"update constraint for oslo.messaging to new release 12.7.3","status":"MERGED","created":"2022-04-27 15:52:14.000000000","updated":"2022-04-27 22:15:51.000000000","submitted":"2022-04-27 22:15:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839586-new-release","meta_rev_id":"d8cca1a9218ffcbf1f8ddae88bf59eb8c7e4afda","_number":839586,"virtual_id_number":839586,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 22:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-27 17:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-27 17:17:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 16:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 17:14:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 17:17:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"af44ffaea11cd70595184f9c79c865fd6ec11982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-27 15:52:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"423f93b04a63c2b684771e86e60b09dd48d09824","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:06:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2d9ccdcc8f5f40178a476bb09fdaf1a2 : FAILURE in 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1c73232c9f92c0c670dd4fd88b5bab45193ca6e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:14:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555d470e74144ea385096a97e7ed2a99 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/359512a74b5d4433b48f9451e50cc03b : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc845317e225486181a2771e7fea6b05 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/273d1293a0b94700a11d4033adf8b7ef : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e11952ae4b0449582164ced6ab59f8a : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8252a0e46d074ca99c28bef987b8f95c : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2789cb4a97a14a888e058295df019b33 : SUCCESS in 9m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a7d510c99bac4ce08f6652e78e69e131 : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ec332076b3ed489cb365c71268ba3190 : SUCCESS in 9m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b0a4fea1fd864c6291d83ab842ba3d11 : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63dd6ebb8e6f4364a5c2bd7b4c178eef : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac5a3f5bd23b40698c49bd4efee9b1b6 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49b257521cff497790140f100b2e99f9 : SUCCESS in 8m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b0861650de554170b4c3a3ef57bdfab1 : SUCCESS in 13m 00s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b30ef7feedd1483a97d1bb20fb86ed51 : SUCCESS in 5m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fe3d68a3ee946129a31792f7f3ddc1c : SUCCESS in 34m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/34d16de412f541988f1175159f4da6e9 : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b07c92688ce242a188b5ae65f12a0410 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64fa6481b8364c078f9895ad2f26765d : SUCCESS in 22m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ae05d31f72554e1f9f944ed9e0153ac8 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f648e081321f4207a7d10873744a2cd5 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/28143042f0cd442a8bfe2b559dddde89 : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ae2f1274f3f4ddf87b4d8f4fdc4e6e1 : SUCCESS in 56m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5116c62e8ac4c2f839b5efa8aff9849 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"846972035c6461cd413ea3203849d9ee90beb5da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-27 17:17:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cfceb22bef167fdd8c6198c4f2027248f55d8b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:19:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab0d0dc869161b3275b950e2135dc8476a62c934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 22:15:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806d5c9a598349529a47f7dbae6f1a8e : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b03ac6d53174413bd73c2dcd191ef02 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fee76e371354013acea2e4ce65d2291 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6ba1a5eac554236ae958ef23af2f046 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd81e0f362bf4150b960aeab5cdcc9bd : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/95067be0719e422fa2ed62b3d712e1b5 : SUCCESS in 8m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/489b59397bb64348812d51cb20e48506 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd41aa5decf04e4e974fb7f08960c641 : SUCCESS in 9m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59e82e2d27a846588d2e99d1fe3e9c43 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c24e65ca9f9e435f81746a46586740d4 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f56ca29a1994a678400b8009e8d11ce : SUCCESS in 8m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0f50cf364ecf4574abdf47f7493a2c88 : SUCCESS in 7m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7488bdccef9a46babec2022181504843 : SUCCESS in 12m 59s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e247d70ff00f48e8a771d31891e8114f : SUCCESS in 5m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aee9eba31b0249129dbb3373832e371b : SUCCESS in 28m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe90ba145b6c46969ff312cee440157b : SUCCESS in 12m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a77ec421405489ea0f6282e55dbe4cd : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06885844e97446dd92a58720fe714de0 : SUCCESS in 18m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bd875cf869d24c1dbb69eb74991d2c66 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1bb95c4d38fe4257b52717515468b525 : SUCCESS in 6m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/33397cbb06944c53ae86ad98caf781dd : SUCCESS in 14m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0db00f24db3475686823b6a6d3249b4 : SUCCESS in 57m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f9ea0993e654ddda565ab7c05184e09 : SUCCESS in 1h 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d8cca1a9218ffcbf1f8ddae88bf59eb8c7e4afda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 22:15:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"320f43f36a86f7741b21f3a2e346c9d48c69899d","revisions":{"320f43f36a86f7741b21f3a2e346c9d48c69899d":{"kind":"REWORK","_number":1,"created":"2022-04-27 15:52:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/839586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/839586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/839586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/839586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/839586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/839586/1"}}},"commit":{"parents":[{"commit":"7e46413c5ab6fba94ddf219b5efe4d52d0dba95c","subject":"update constraint for ovsdbapp to new release 1.9.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e46413c5ab6fba94ddf219b5efe4d52d0dba95c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:52:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:52:12.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.7.3","message":"update constraint for oslo.messaging to new release 12.7.3\n\nmeta: version: 12.7.3\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I61e3c88d31680709b71ff8226fca29a6e2695b94\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I84b8e11c602b002896c01812381772f815486ace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/320f43f36a86f7741b21f3a2e346c9d48c69899d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/320f43f36a86f7741b21f3a2e346c9d48c69899d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
