)]}'
{"id":"openstack%2Frequirements~839587","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I68bf128274078653d4f072c58cc999d96b0be1b7","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I68bf128274078653d4f072c58cc999d96b0be1b7","subject":"update constraint for oslo.utils to new release 4.8.2","status":"MERGED","created":"2022-04-27 15:52:45.000000000","updated":"2022-04-27 22:18:56.000000000","submitted":"2022-04-27 22:18:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839587-new-release","meta_rev_id":"3bebfe7715f01dbfddcc7d4a86c1fe4c9a68665a","_number":839587,"virtual_id_number":839587,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 22:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-27 17:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-27 17:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 16:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 17:17:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-27 17:24:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a3f5575b24f5cfcdd6804fea869e556e64cc5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-27 15:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d42be826a83f217572f2fc9485dd1de087807b8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:08:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/427394f285384f7fb7228de0a352ea89 : FAILURE in 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"99181403b38419fd7dbc1b4d59257c88c49e6225","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-27 17:17:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f12f94e38c0049e07dabd2c2e4d37b85db20732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:24:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f033082a0d2465fab9e89387d1db0de : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05e5a8d9ce02403380f2cdb5e1fe3290 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e02ceac0ff874c8191500570dd038006 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e4faa3bbfef4c2aa7c37a7d3aabbe58 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb3f04e6f0b04c10896b6aa1dd835cbd : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fbb02933f8a1407cb428a79587123537 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/594b566a6db7428880eeb7ef35eaabb0 : SUCCESS in 9m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/690b00f819db43d297c9504bddff34d3 : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/20d1c4d3fa994267b620cd3554c5f55c : SUCCESS in 10m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32674d2948814973b809811039e84aee : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e0fce5e096964b449c8dc3ce3fa95abe : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8367be8993c64fe283e9c02aa3f0269e : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8079b18e3be54ab8866104c1def3f772 : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/98a78b14c70444a399267cc3e66822d5 : SUCCESS in 14m 01s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/71dc563d8224425f9d2fb66c4a5dc474 : SUCCESS in 5m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d6fbc108ee94b14ac184726e42606d1 : SUCCESS in 29m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ad25a60802cb42d0b30b298ae70e8023 : SUCCESS in 12m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1214ad3c4e974de8bc6d773289ffef27 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b73594d163944254bd32ef2e7db48973 : SUCCESS in 19m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9ff320d1a57440999f0c04845141397 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3176a6c1eadb4f8dbe2b40d8df1e9d79 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1f43ef4db1b144acbbbcd8ad02f63b0d : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffe7d04b940541b2a60f3bd66871a156 : SUCCESS in 53m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e871e1bdf2849f4acb4d72f6061ae79 : SUCCESS in 1h 25m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7b9cc89841711c8228df8b80ec8b5d12cfbf3b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:25:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"339545b447eb4e2f72448c72ad6129bfc089414b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 22:18:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cf8fcb1e6354d02a791e978bf99840f : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93033ae225f24099a68afd87e85e3e44 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afa618196bc74bebaa1a90b0919d5a49 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cf786460bcd45b1942ae9e7f8fa6cd2 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3befa281ff51473e8901e16410904eb6 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f882e80a80d044b18d706396e2914b3b : SUCCESS in 8m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dec9b563cd954db6a89d259fbbedd15c : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4373c147b23343cfa1346b01d578dfef : SUCCESS in 9m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5eaf0d6ede348ef841b5245b968596d : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4bfa9bbf2e04600871e039629bea236 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81625e3a94fc443dbe38e837169c5f8e : SUCCESS in 8m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c610dab2836645bf8028990a8afd67dd : SUCCESS in 8m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1200c520ca54aaca86e14c4d3a083f4 : SUCCESS in 33m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e27adfeebd254edf99b6361491ce951c : SUCCESS in 4m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/19f9ae6f2c3a4ef7af18663e5bcf88d9 : SUCCESS in 29m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/45562188f7e34d73b19ad59bb676240f : SUCCESS in 12m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae5bea8fa8744d87aae28850b5f2672f : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd3f239a84d24fc4902a72df484ae4e8 : SUCCESS in 18m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a9cf7f7429e4b6bba210e0453c29910 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c6c33edb8d44c1999cc5d5a892f3528 : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5173af789f2f40a7aea52c97dea63d5e : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ba729aae12948acaa1466cca22ff58d : SUCCESS in 1h 02m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dcdaa2c79194c38aae27b9d36722630 : SUCCESS in 1h 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3bebfe7715f01dbfddcc7d4a86c1fe4c9a68665a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 22:18:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00fe0e1f1e6583bc014b561df1c4b0391d07ea62","revisions":{"00fe0e1f1e6583bc014b561df1c4b0391d07ea62":{"kind":"REWORK","_number":1,"created":"2022-04-27 15:52:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/839587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/839587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/839587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/839587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/839587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/839587/1"}}},"commit":{"parents":[{"commit":"7e46413c5ab6fba94ddf219b5efe4d52d0dba95c","subject":"update constraint for ovsdbapp to new release 1.9.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e46413c5ab6fba94ddf219b5efe4d52d0dba95c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:52:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-27 15:52:44.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 4.8.2","message":"update constraint for oslo.utils to new release 4.8.2\n\nmeta: version: 4.8.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I61e3c88d31680709b71ff8226fca29a6e2695b94\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I68bf128274078653d4f072c58cc999d96b0be1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00fe0e1f1e6583bc014b561df1c4b0391d07ea62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00fe0e1f1e6583bc014b561df1c4b0391d07ea62"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
