)]}'
{"id":"openstack%2Frequirements~839765","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I151cc048556911bc014d18f13fc3c69a616ec1c7","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I151cc048556911bc014d18f13fc3c69a616ec1c7","subject":"update constraint for castellan to new release 3.10.2","status":"MERGED","created":"2022-04-28 13:55:02.000000000","updated":"2022-04-29 01:28:16.000000000","submitted":"2022-04-29 01:28:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839765-new-release","meta_rev_id":"0283829db486c6cbb2007645057cc4b086fd74ee","_number":839765,"virtual_id_number":839765,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 01:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-28 18:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-28 18:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 14:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-28 15:19:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 18:31:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e9c907f7abfcc0eae5e981209df5f687e3d98863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-28 13:55:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b544f6419122d2a2e64ecfe02be923d5581fe928","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 14:15:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4e9956b5552a4af69be26e4d7c0b7c47 : FAILURE in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9cf8679b8bd43065de319526b25da8c5cf82edd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 15:19:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/529b9e1f71854c99acbd06e2bb0ec1f8 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02aca520f9e14186bfbe2348bcffc5fe : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab4edde6dacf4606a6c2114fc4c80040 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/234e1764095b4b7abfa6d9891fd3d793 : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62793b5d9b1c42c094944eaef4f3fc69 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6164e49cf9f14efda1af1c06527c8933 : FAILURE in 5m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59ff9b84df2a44619c3406df49ed7250 : SUCCESS in 9m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e7ddef3c859b46e2a0d7b64c34de0a73 : FAILURE in 3m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/443995dcf749404b8117642385ea9b3b : SUCCESS in 10m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1d9bd720a2e749bc88a73d290c13ffc6 : SUCCESS in 5m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e2f745bab08043e9bdd1163da2154dcf : FAILURE in 4m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d1d59cdcd9de48e6aabf93cb0740f074 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f17a95fc1f0249f1948f9b84ea3053ed : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6a27cc0a60f54c58b3bd24f702bcf213 : SUCCESS in 7m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3b16b397b5fc45bf99c5a3016c5cd95c : SUCCESS in 10m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/28be132c751a4c368d4786215b377f3e : SUCCESS in 4m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/06127c961eb246bcb421a82170879c37 : SUCCESS in 10m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/912a56fa3f9348f4955d20fcbaa72335 : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3f10a8e8e6ee4ffe9f9878ba171e04ca : SUCCESS in 22m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b25fd555c41f4bcf8d27732a6b374744 : SUCCESS in 13m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/80b8a4b8b9a347889a2f1b9e2b715d63 : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3bf8d629335a41148d6dc8485d36f4e1 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/485e9b66739841f5bd8a4b05783350b9 : FAILURE in 5m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b1b9ec423bd149eca787ff015924dab9 : SUCCESS in 7m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e2bf48447ae943469126ffea9e8aca29 : FAILURE in 3m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74ff1d4b8bd54d4eb18b7a1c24b60d62 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/58583c5e305d4717a1a06e2396e3838a : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68881df4572945f398a1bceccdae41a5 : SUCCESS in 48m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4e26250e117444d824908c5b98a4228 : SUCCESS in 1h 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b76174b5d4db6def55064accb7fbf937cf324e3b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-28 18:31:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9d42ae365aeb2bf31acaf8d45e5e51e94cace907","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:23:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/72eaf16531204eeeaa477b7a8c8ff931 : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"543a5486c07d0941b3363d8d753c2f7b7fb8736a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:06:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88be0761a0cd4b09a4cea951cf97521c : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c56a1be0f9d4a669b1d88aff38cc0e6 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f858059a28e54a58aabf383ba66ff39d : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba875fcd0f77491da48db427c4cad59b : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c700d3fb05a347b4a3ac30fa84b601ad : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4dc6e61ae56b4a22adc265c8349b1dd0 : SUCCESS in 9m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f9bd49e4312441fc91f8b79e792af447 : SUCCESS in 9m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b7450ba0da474bd98c6ef0c3e90b6b4a : SUCCESS in 5m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5a7a868dcec24be78dd218769da57adc : SUCCESS in 10m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c46070b0a27f410f888b55f3b82d176a : SUCCESS in 6m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/069e26c5b4be438db1ce13b3ff069090 : SUCCESS in 7m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cca5d1ecd14f4b71a33d79a9a31cd6e6 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99cf4a43759f48f18c1954d0ce9bc6f6 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c2f7073090ad40f4820a1ccf284e07a3 : SUCCESS in 8m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a0fcdb2b0282436ebf7b0de022d2850b : SUCCESS in 12m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/05cdf041bf144e23af84c92452b1d0c0 : SUCCESS in 5m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/57cf0ea401654bf6afb58a0bbcfdccb1 : SUCCESS in 9m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a212d71493224d68b2f13e3ad976d3b1 : SUCCESS in 6m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b38485d6853c44529523061abd91ed4f : SUCCESS in 21m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/96becbf70ee94ea7964e9a2a9bb2ab0d : SUCCESS in 12m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/650bb4d5625848d28aa56a0ee565ba29 : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf49732a26c941ab8f0430b97e3093ab : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1fdb66c7208413bb89ef5fcb4caedc3 : SUCCESS in 22m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3cf403af14de4cbe95e10f3d294a7d4e : SUCCESS in 6m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/41c8f16b30bd4bffb96dd3fe54356e98 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e086637460a14089806a1f1cde2738e9 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/30fb98c3d4e44997b8cf626bdd49ada9 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfbb080159eb4e50aed75982fd93bf95 : SUCCESS in 51m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4004ac246a094e7a83c2ac106ad07aef : SUCCESS in 1h 29m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8d6ba2effed356e86d1d7c7010c58dc028ee8c4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:08:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b586a0bd555d04306e9833e2c2b6eaa1bc6b45a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 01:28:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ebab899c7343699cf26a4a7f959faa : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31d3c92e7e4c494a85a28d16f6ca2838 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/848ecb83f9fd4beea2b462fb8577b47a : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7620a1fdc1f94eeeafff62ad7735e610 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6557bbb15f849eab9de00eeeb650e38 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6955e1dd05d49f2a417627beeb4cd69 : SUCCESS in 7m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a13b9c4d03b5490c9e2ae2d31fc5cdc5 : SUCCESS in 9m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/64d593dd4dc9411bac16391586358653 : SUCCESS in 11m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e531cc1d5cbe4aa6bcf726c6640e9933 : SUCCESS in 4m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/da5ab3492e064498b4a86ffcdc0d1f8c : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e5123e3560da47b9bee5cda1615b0c54 : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef5502cec1db42f5b3e245fef43db0bc : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/622919d9f2b4413983b3d35e54f55b4c : SUCCESS in 8m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/afcebd5fd4214e3da6059938be14fff8 : SUCCESS in 11m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/52311b9c2151450f83f58cff44f97a6a : SUCCESS in 4m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b6f486ac41934e6eae0f9ba2155b3cff : SUCCESS in 12m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/27de2b2015ba4835873ab72d13903887 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/524a6be150674e239532dfd8c6d0f977 : SUCCESS in 20m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/05d333dd371246fbbdf7398f7b8c2748 : SUCCESS in 12m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ee7106d1ff0549a58702c27d7a3cf402 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bca2782147db4be9961e0f5d04ce3d8f : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd21bc0b2abc426486e4b27c9dbdd4b3 : SUCCESS in 18m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ec583519453340cfaba6948b299125f1 : SUCCESS in 5m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8d8cae81af0e4cb1bb4b8f7abedf5e19 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/385fdb915d01468d82d29d51b5a62bbf : SUCCESS in 6m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5eb107243018486a840818c1dc5aa5fe : SUCCESS in 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d328a54f1ded4ee6b0079bafac58284a : SUCCESS in 39m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fceb102092584d719cacc05e56e4e395 : SUCCESS in 1h 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0283829db486c6cbb2007645057cc4b086fd74ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 01:28:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"760bc51a174db84841cfbc2fd09f0ef1945b3a97","revisions":{"760bc51a174db84841cfbc2fd09f0ef1945b3a97":{"kind":"REWORK","_number":1,"created":"2022-04-28 13:55:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/839765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/839765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/839765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/839765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/839765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/839765/1"}}},"commit":{"parents":[{"commit":"5e5970c842d2a88c832279e253e815918c6777f0","subject":"update constraint for ovsdbapp to new release 1.15.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e5970c842d2a88c832279e253e815918c6777f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-28 13:55:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-28 13:55:01.000000000","tz":0},"subject":"update constraint for castellan to new release 3.10.2","message":"update constraint for castellan to new release 3.10.2\n\nmeta: version: 3.10.2\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib493c2bd32c1bcef28d971724442abc2a2690b66\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I151cc048556911bc014d18f13fc3c69a616ec1c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/760bc51a174db84841cfbc2fd09f0ef1945b3a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/760bc51a174db84841cfbc2fd09f0ef1945b3a97"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
