)]}'
{"id":"openstack%2Frequirements~839882","triplet_id":"openstack%2Frequirements~master~I03767ec8e9e682ffc54e58780515e36c1db23811","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I03767ec8e9e682ffc54e58780515e36c1db23811","subject":"update constraint for oslo.policy to new release 3.12.1","status":"MERGED","created":"2022-04-29 09:24:43.000000000","updated":"2022-04-29 16:40:23.000000000","submitted":"2022-04-29 16:38:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839882-new-release","meta_rev_id":"1dad645fdccb8856740ee3a1a464e49a660a7ad1","_number":839882,"virtual_id_number":839882,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 16:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-29 13:18:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-04-29 09:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2022-04-29 13:18:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 09:28:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-29 09:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-29 10:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 13:18:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"afbfcec35767efbc57552b897aca7baf5ac77a9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-29 09:24:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b972ef124eea0e8906cd65f1bf54a879c70467a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-29 09:28:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6b865e862347f4748dbf3cf3604fe58c20ef6c6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 09:43:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6fcaa6a5189c422db87303a075ef8e76 : FAILURE in 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2e52065b9eddf808810204e7fbe02310effbb08e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 10:56:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94eeaa457be34ca28ccbf6c6c0296404 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40c733378cd4e31b6d9fdadbf07d4de : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bb4b274135d44baac46d7b005dfafb6 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b502664259894fa4963e4a815e848c58 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53a5d578ae4047c3bcf1c58d9427e4a4 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/440bf78e10cd4ee09f1e251b9df01230 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08a0112d03554ec0a1164f8d2094edc6 : SUCCESS in 3m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17214eab68d2449bb4ba3594bfce8822 : SUCCESS in 11m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4603c86679bd469eabe8f5923a243a76 : SUCCESS in 15m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/17b4e4a9855d43cda565a00c7e58449a : FAILURE in 2m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/be96a4448cad416cbf77b5c1e2252b94 : SUCCESS in 11m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f6163dcfb6ac4b6daf46e3e0e94e1a83 : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f9e7d04dc8e04093a3566a3f7255b61a : SUCCESS in 6m 48s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/52d19903aee14777a91ba348d35d9bb5 : SUCCESS in 8m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7ed2204f0d134378a82f10ba2f6e020e : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39ae78926a6247e4aaaa511d559bcea3 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4631291c14445d3a5df5e74e00d422f : SUCCESS in 7m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6f4e1d56d1dd4442961936ef8b9a783f : SUCCESS in 12m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/135ba5271a754b65afae42c4bbcfe937 : SUCCESS in 5m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/364eabb5666f4a3a8b5ff285ae13f926 : SUCCESS in 10m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c06ffb12bd834039823207710785f882 : SUCCESS in 4m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/74275bb1736241fc9a94aa2db48f3dfc : SUCCESS in 21m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/baa8561e937e435e8a995c2c7f2151f6 : SUCCESS in 14m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0a35792235314d21bac580aca41abc6e : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b64ae22fc4d24387982853ef69590ae3 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66dddfb163564eb09ec5ae5f054c9f4c : SUCCESS in 18m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7ee82f2c7e954158b54b83c7257f40b8 : SUCCESS in 6m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/84be39cbf4ce4f62a9a22dc26d590fa3 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0fd6869849a415db6eb6255045f7947 : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90dca32786d94dc692d7edaf679253d9 : TIMED_OUT in 41m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c37413ff130490394b455c4b32c697c : SUCCESS in 56m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/608144e75a6c4928b1dab606caefe400 : SUCCESS in 1h 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d11bd868054e66e9f3a2f3bf5597ae0d4d998acc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-29 12:28:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba8c612ff24a43133a2eb564d72f43b2b78f234e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:48:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ad9aeb0373c24c848b8e466cbe0bb768 : FAILURE in 18m 25s","accounts_in_message":[],"_revision_number":1},{"id":"59e48d2473341feab1d1c0790da2ab715a4a7eb2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-29 13:18:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a76db0742a87e2f0bea475bd7911a3f85ee77f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:10:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9e0e3fa45c4f49b5411b9761a4a467 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f412b9fd3e864dbe939d936ca772aa82 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce85f0811d2a41ac8b104ae417882fe0 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65ef4c3b7ed940a48e6a3195c0239ab3 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccc9a1ed75eb430aa1895e62e474c72d : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d0b7ad81939492ca1d624a464fb3e0b : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9970b90b4db7458f934a61c9a8a046b8 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/812e764cd1064835931ac7a6b25c70e5 : SUCCESS in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ee7b6d85664d4c378bffa038bb2e9d93 : SUCCESS in 15m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b2d66cf44c1348f5bcaa107f1408cf40 : SUCCESS in 5m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8b5482211ae54ce5b93da1993692f37f : SUCCESS in 12m 14s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fdd0f70a64464d5fa7db74eda1405add : SUCCESS in 4m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fa6cced46d414c20bb9f120f8bafb63c : SUCCESS in 5m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/04b2054d7d164b8ca358d579bdaa582e : SUCCESS in 9m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d275ce4329664dae8d2e72700bb993a4 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4740d1217d794d49a8e4a8be22d2f8cc : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1a1079e0199244b291e5e7b01a36d175 : SUCCESS in 8m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/90a711d7b19447f8ae1c2b15340a054a : SUCCESS in 12m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6ed913e06d5e427f85c9c67d47ff7498 : SUCCESS in 5m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f7d4fc1ae5474c74bae8654bbeef0531 : SUCCESS in 9m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/da973255de2f4bc98642a5fe282c1a86 : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c66b709987274ed5bc8bc09711f32e55 : SUCCESS in 26m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1338b8961995439e9e59c62b533e24dc : SUCCESS in 18m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cd4f8cfbe5d3435dbda928d6e87e919a : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7cdd51bf5464cf8ad08962c9f923102 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f68f74d805c41f9ba4bcc8ca7e9e2eb : SUCCESS in 21m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/33a6eb80426447279d65d2858ca5bf43 : SUCCESS in 8m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/959848ae3cb048cc97c54e429b78575d : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e85e709206843dda4be2d07ec0f924a : SUCCESS in 8m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/994c3daf88764953a22ca4e35618125e : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55c31cc9c9a84e559436b622121ac41c : SUCCESS in 53m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95952bbc98614a0e89ab4479f4c2b0e4 : SUCCESS in 1h 34m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c88f7e04ed5287273634c4f3109238b1d2bd202a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 14:11:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a61d387c1c9663fde15afe7712b75418e85c2b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:38:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb5336bc6394657aee6ddb8d5e9db67 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9602eb49fdc44f458f3921c27d88ae36 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/731a5390d6484d338284d39e4e0285e1 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ef6d63f9c0148c583c361d1253db2c3 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e0035e9a457443b86ed97b6cc368259 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb2785349fcb402ca3ef9a526f71899a : SUCCESS in 5m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/71974d91fb4c4a73aeeede62fd27c7c3 : SUCCESS in 12m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/12a9ef58f4c94b41a720ff648de57f48 : SUCCESS in 18m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0a26d2dd3c474b27b34229d3867f71a6 : SUCCESS in 12m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ed229414dd844ed49354aa7d38554448 : SUCCESS in 7m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9485727c8a2f459fbe0c50215d8879ca : SUCCESS in 7m 45s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7ca41c2fd499444c985a1d47b0406d6d : SUCCESS in 9m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3002f8f2f3d44cb6b73e089265c4aa13 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/984b9157ca9a4d8f888ecf5b9ecb2572 : SUCCESS in 9m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8b1a79b6c02942cfa82933e1b7e290a7 : SUCCESS in 9m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/44e658204bb14787a7804d5f2632545b : SUCCESS in 18m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8ab434a7239049209f0d7c733ac4252f : SUCCESS in 6m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e785720232f5424f8619942f3e7af22b : SUCCESS in 10m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f128588f8b4d4ddfb2bfffb244b0df50 : SUCCESS in 7m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/631c289fd8f94cc1aa5c76c6194bf940 : SUCCESS in 26m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3471fe23e1344e4ab82a874f669ec03e : SUCCESS in 14m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6b87bd24096f446685abe46d6b60781e : SUCCESS in 7m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4886055fef784b9cbcfcce7754e3d73d : SUCCESS in 6m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3dca98663ed848de9878d547bf68e2e4 : SUCCESS in 21m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4f248e9f9f0849938f8483d6b1c1038b : SUCCESS in 9m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2288bb31c8e24492a21eb1ed7b2a7d07 : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5a867b1690e485f815988b1b7bb94ba : SUCCESS in 9m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d1c36708159342509a4ccc57afe978ed : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e175acaf1f44d7dbb5124c229fd3d59 : SUCCESS in 48m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c114f1b52824a3c92364c70a7bcc56d : SUCCESS in 1h 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0d52b1a309beb9e9faabe7fa0ec092f65f06d564","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:38:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1dad645fdccb8856740ee3a1a464e49a660a7ad1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:40:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb984a47b94f4eee8a645f0dcf5a9bcc : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ba7493ce35c83a4543eeabd2b3680961ac4bd68","revisions":{"0ba7493ce35c83a4543eeabd2b3680961ac4bd68":{"kind":"REWORK","_number":1,"created":"2022-04-29 09:24:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/839882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/839882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/839882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/839882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/839882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/839882/1"}}},"commit":{"parents":[{"commit":"e767228e60bd028306c459a8fb1efebcbb969e35","subject":"Merge \"update constraint for castellan to new release 3.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e767228e60bd028306c459a8fb1efebcbb969e35"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-29 09:24:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-29 09:24:42.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.12.1","message":"update constraint for oslo.policy to new release 3.12.1\n\nmeta: version: 3.12.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I13db2a6b5d7ea4026a17de543d425d1b4431c8ee\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Code-Review+1: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+1: Balazs Gibizer \u003cgibizer@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I03767ec8e9e682ffc54e58780515e36c1db23811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ba7493ce35c83a4543eeabd2b3680961ac4bd68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ba7493ce35c83a4543eeabd2b3680961ac4bd68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
