)]}'
{"id":"openstack%2Frequirements~839908","triplet_id":"openstack%2Frequirements~stable%2Fxena~Ie9eb2de0efed8576c49398cef5b1344859876219","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"Ie9eb2de0efed8576c49398cef5b1344859876219","subject":"update constraint for horizon to new release 20.1.2","status":"MERGED","created":"2022-04-29 11:49:19.000000000","updated":"2022-04-29 16:26:29.000000000","submitted":"2022-04-29 16:26:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839908-new-release","meta_rev_id":"3f72f1a313197ab0922a9208d6e3c6a1eedb644d","_number":839908,"virtual_id_number":839908,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 16:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-29 13:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-04-29 12:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2022-04-29 13:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 12:11:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-29 12:27:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-29 13:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 13:19:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"6b7ba64fd69482043e0d52dce1a46303d9783b08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-04-29 11:49:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13031b2f2389fcf24528d2d17fbc128ce3169f89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:11:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b854a90671fd4d4499f90c15f7c00c26 : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"12632ced2c86cc16e0b7907216200efd112a18ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-29 12:27:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f5cf67641839463be15408b05f16ba9b499c81a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 13:08:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f740251f8f2472fb4ffa68dea3642e0 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a8bb26df104473385fe0cfd559c9aaa : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75eaa6970e694d4891f7dc7d1898191b : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a350d75ac61e4ff5a14f64bae555a5e5 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24c67be0effb498c8bc63f167e5f661a : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b1203d05e97431da21937952bf272ac : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0002b3ee38ed473891ea1d5445f959ab : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10ea2ee2e8184853830713e551f782f0 : SUCCESS in 8m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1e73ff7e86b34a4fb9c1c6259c46d92c : SUCCESS in 5m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f9ef86942284445da03ba51092b345c5 : SUCCESS in 9m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4be061e61cd54c82ba4dce229e374f98 : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/168a7db33984455ea4e5eb1cf289483c : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f86458a80b64442db59b8cea51a111b7 : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e6a525fdb83c4004b45cade8b62fe487 : SUCCESS in 8m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/79693a9ce09b45dd98e5691ab9ed123c : SUCCESS in 15m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c221668001dd410bb2806ffd230f76d8 : SUCCESS in 9m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2aba3a0d311a4df5bd2407f027f4ae04 : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/534cfcf9235e498487fc666b566f5426 : SUCCESS in 22m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/94915702178448b0bbd86c5be8d668e0 : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e6e8cbc4acf8432e9c5b69efb06c3e8e : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/658b30b1ee334342bc62dcf7a4c4df12 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d97b61250684f8d878d09a7dd70febb : SUCCESS in 18m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a86e6afbefda4a66b9a2ceeff4e8fe70 : SUCCESS in 5m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/51d959c6e83340b7b8fdf163074161f4 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9a2035179c44528a99af68036565de7 : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d68d84943ee149d38b6af954e74278a6 : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbf2eab2f3594f8d99ec3305faf47c20 : SUCCESS in 47m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e102c68426524d7a84be04c5ea3cb010 : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"297965b1b3bbd0f0ed4c91429e33ed8c150d94f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-29 13:19:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3892cb4101daeef132e6c92b8de1057a21539fdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 13:20:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79e349b0ca3d17ce0d52bc1c9016e6d75481ac77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:26:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ee85242ee94f71b539d3f26ac6939e : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b11ba842cb34ff088b147ac0e501841 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97ee1b2d85784809b610c6c3dede60d8 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f62ef00fc27c42acbf0fba9a6fa9d62b : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/982270a5b48945b5b5b09d55a08c2b51 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70331a483645453aaaf9270c1fcd0080 : SUCCESS in 9m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ca21de2d5954beaa9f5145b5f5c7445 : SUCCESS in 8m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3532cffeb7964f86a729e46742ff6c0d : SUCCESS in 9m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d419e5b40fc94391988585a9e44b5736 : SUCCESS in 6m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f5d5556f9ea940a5a2a8163d9e7292fd : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d668943178c24d92890940639a20dde1 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/05c9a66e554d4f228453976d6bb819ef : SUCCESS in 8m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5bdab40cf5c843d0b5dd8b0630bb496b : SUCCESS in 15m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/68ef579be60e4d878a144a220730c73f : SUCCESS in 12m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1d3c8bf3c69042c9adf2daa201ced2a9 : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/65190c1f330746ea8783d80cdaeed801 : SUCCESS in 26m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ac6bee1de3734ae292c2c82326dd35bf : SUCCESS in 12m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8645a755cd5e4ae48cc21e0b5120a12a : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2e8a6ffb1a464dfeb2c5d8749fed786b : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95452ec947d34bb8b2bcffe72c101ff3 : SUCCESS in 23m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/58a770d66f11430ea652c477a1c74891 : SUCCESS in 7m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1d715e1239d0441c852e51af501d3acf : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd2e4440d2314c3db0206a0590abc437 : SUCCESS in 8m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1e226c668f01408dbe8a2055a1f54505 : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f06ee6e668e04d5f9487f8a3e8a5b1de : SUCCESS in 45m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/118024af1aab44faa8d449684ec1ab6d : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3f72f1a313197ab0922a9208d6e3c6a1eedb644d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:26:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b364534b1405ec4be76e79729ccb0eb5bc8f3e4","revisions":{"4b364534b1405ec4be76e79729ccb0eb5bc8f3e4":{"kind":"REWORK","_number":1,"created":"2022-04-29 11:49:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/839908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/839908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/839908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/839908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/839908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/839908/1"}}},"commit":{"parents":[{"commit":"876f72808eff679069f2cea6fa0452a8536c0f61","subject":"update constraint for ovsdbapp to new release 1.12.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/876f72808eff679069f2cea6fa0452a8536c0f61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-29 11:49:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-04-29 11:49:18.000000000","tz":0},"subject":"update constraint for horizon to new release 20.1.2","message":"update constraint for horizon to new release 20.1.2\n\nmeta: version: 20.1.2\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta: release:Commit: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta: release:Change-Id: Ibb821d6b2a44861ff15bff65bcbb91b0b0331565\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie9eb2de0efed8576c49398cef5b1344859876219\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b364534b1405ec4be76e79729ccb0eb5bc8f3e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b364534b1405ec4be76e79729ccb0eb5bc8f3e4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
