)]}'
{"id":"openstack%2Frequirements~840109","triplet_id":"openstack%2Frequirements~master~I421cc2fd4404a75edd8b51521f84170cdf302b39","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I421cc2fd4404a75edd8b51521f84170cdf302b39","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-05-01 02:39:31.000000000","updated":"2022-05-01 03:32:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c69602d361ae767204008662ea048be54e3fa92","_number":840109,"virtual_id_number":840109,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 02:55:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"71c7242597d2d7bf8b8df846a14668fa1e4c2b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-01 02:39:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41f3e5c63a42fecbb9fa532eedf6278f734eb35b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 02:55:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/91f085f7ed7e43ca9462347b165d5d5a : FAILURE in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1c69602d361ae767204008662ea048be54e3fa92","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-01 03:32:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddbb076168e246534174ed3cdf1b4f3887d231c4","revisions":{"ddbb076168e246534174ed3cdf1b4f3887d231c4":{"kind":"REWORK","_number":1,"created":"2022-05-01 02:39:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/840109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/840109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/840109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/840109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/840109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/840109/1"}}},"commit":{"parents":[{"commit":"d058b62cd06f80034aa2853e46e01c3052ada7b2","subject":"Merge \"update constraint for oslo.privsep to new release 2.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d058b62cd06f80034aa2853e46e01c3052ada7b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-01 02:39:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-01 02:39:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I421cc2fd4404a75edd8b51521f84170cdf302b39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ddbb076168e246534174ed3cdf1b4f3887d231c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ddbb076168e246534174ed3cdf1b4f3887d231c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
