)]}'
{"id":"openstack%2Frequirements~840562","triplet_id":"openstack%2Frequirements~master~Ic6af283d1b3dba24e898f7fcf25d52e2bd2a9959","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ic6af283d1b3dba24e898f7fcf25d52e2bd2a9959","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-05-05 02:43:18.000000000","updated":"2022-05-05 04:20:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5aff6b2d2315a732a85ac32193ea044c0530c86f","_number":840562,"virtual_id_number":840562,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-05 04:11:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 03:02:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 04:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d72a5eedd14d347cd8241580cbe508fe022e2f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-05 02:43:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8f42413305b307126e4e42d708a510564144308","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 03:02:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/20c1dcedda97415cb1efc8be2ba94806 : FAILURE in 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"62d60e14027927d1a8502c69439f3f0a73fab51d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 04:11:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640384d7ccba4bbda8bf15a6bc2d962e : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f5ad3d59d8a49f582f93f3ecafaf818 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8427117f8654787ad89a9b570561806 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f124b66fcfd4b2dbbaa82c37d8291e9 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeeceaf876064057ba94e5d6e01fdcb0 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5defe4a795094fd189941002269f4323 : FAILURE in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cfc86adaabbf4d019dd191e408e39fa1 : SUCCESS in 4m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc4ebe92ebc9407290b71fe92a4a7902 : FAILURE in 4m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c0f05f4a2a0141f986ab3157b8a3b9c5 : FAILURE in 4m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/acd0d48d402b48e5bfc2f8dbc7043682 : FAILURE in 5m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/12a55950c2034a71bf80f72c996daff1 : SUCCESS in 11m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e0fe61f2da554ac9b597482de0a82255 : SUCCESS in 4m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5690b40ac8094d7c93567151ec42bf9f : SUCCESS in 5m 54s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1098759054664b43ae29daa6682147c5 : FAILURE in 7m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3ec9d4e6d1614b23a15b024b8504f6e1 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff84e2baf28f41adb1f2b33fe285085b : FAILURE in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/42a3c8ece2ad4999986410fb4e9ea5cd : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/80da5318bde845939f7b1ff7c13b451c : SUCCESS in 13m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/aa1e391e01aa40a0bc57b6a274f71f8c : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f1142c6c4c0f4e70a3d8a3db4f00ec79 : SUCCESS in 11m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fb8075944b844f8f849c982a51f612d4 : SUCCESS in 4m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/61391f45cc6146d98b7f9394d3f74f40 : SUCCESS in 21m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6f6335ec02bd4eb898e09295b9b5a3d3 : SUCCESS in 10m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7b5161f89b5f4e1bab3198189cf490f1 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6ba05d2d6b714aa69c4e7d899f48a4db : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e951048b664c40d1a4295b8fb4b3c960 : SUCCESS in 18m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f1925349c7e948f5904e3ded0a4edaae : SUCCESS in 5m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dc25ea6470b24e409bc8997a3f139bf6 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04ad96c6647c4a119ca84dbbad5bf719 : SUCCESS in 6m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/280a62cc4fc54eb2999ee82b77a1ea90 : SUCCESS in 9m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3050f6d4a64446318b11185aef7d3e4b : SUCCESS in 37m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/662e64e8c8414ecea398600b29a15364 : FAILURE in 1h 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5aff6b2d2315a732a85ac32193ea044c0530c86f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-05 04:20:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f72fbdcf7d55b4ee067d7f1244bbf4591924f61","revisions":{"1f72fbdcf7d55b4ee067d7f1244bbf4591924f61":{"kind":"REWORK","_number":1,"created":"2022-05-05 02:43:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/840562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/840562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/840562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/840562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/840562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/840562/1"}}},"commit":{"parents":[{"commit":"3c70957648ac44105a755990bde97847d4225644","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c70957648ac44105a755990bde97847d4225644"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 02:43:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 02:43:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic6af283d1b3dba24e898f7fcf25d52e2bd2a9959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f72fbdcf7d55b4ee067d7f1244bbf4591924f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f72fbdcf7d55b4ee067d7f1244bbf4591924f61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
