)]}'
{"id":"openstack%2Frequirements~840622","triplet_id":"openstack%2Frequirements~master~Ib5fde11388340c73da79f8faac9297d2a9851e7f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib5fde11388340c73da79f8faac9297d2a9851e7f","subject":"update constraint for pbr to new release 5.9.0","status":"MERGED","created":"2022-05-05 09:57:24.000000000","updated":"2022-05-05 17:20:41.000000000","submitted":"2022-05-05 17:18:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840622-new-release","meta_rev_id":"db82b7ca080f61b20b462d372ae8935927494c86","_number":840622,"virtual_id_number":840622,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 17:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-05 15:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-05 13:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-05 15:33:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 10:15:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 11:17:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 13:19:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-05 15:33:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8d5d03d778881f066c9be1cd8d4ce74d227abd9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-05 09:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31dddea3cdc6d56622d5ead73c2ada638255c9ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 10:15:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a796116b43d14258a3854cfcf00d1f2b : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e777f945ea5fe4df36c67bf43cf90ed017e4bb8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 11:17:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17830a6a325343ad8a38009f4ec17755 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94df86c8ab84a08816840c0b70ebfe6 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30099af1b40143f3a68eb8640c28560b : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40176edc1189418cbf5748304d96754c : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7672557008ca4be0b651257194516c95 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/248e0a6eb3e6426b86380b6dc92ccc10 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf1f5847c4b34ef6985c351f92c82804 : SUCCESS in 4m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28443fc7ed69489cb655af5ce21d2c1a : SUCCESS in 9m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1803a2f4a21e4b5d9c11bc64bf8d58b7 : SUCCESS in 12m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f43bb51250404e51ab11ce6f9360cfb6 : SUCCESS in 5m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0b2257e63c6e49e39fb6d77a179cdbde : SUCCESS in 11m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0fe7c5f22b2041d4aad5ac945090bb81 : SUCCESS in 4m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/81e3f7ef68a14edd8808131e54e37f7d : SUCCESS in 5m 17s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bb40d8b2bc414112b20ea14d144ba2a1 : SUCCESS in 9m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8c1624e8ed494e1a988da28736201b7b : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3bf826480664131a24ff7364a2dfee3 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/051af610400c4e7abc5afbe1695bf03f : SUCCESS in 7m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a65fe67ad6144a6a8975ae97f988254e : SUCCESS in 12m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e37d7200fac94d4db09ca50582a7dbba : SUCCESS in 4m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/02ba99c4c02245c8a16875efb34b384f : SUCCESS in 9m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e8db40041cf74cc4967aba149fa14309 : SUCCESS in 4m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c5ef81fade3c48ddb930d46b7e53993c : SUCCESS in 27m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6721bd4b19b24e1b82179edaa289de05 : SUCCESS in 10m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8553867e91c34030b133ffeea730db32 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/589e3fbe9d134bfbb7da23ff60b45633 : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99c8c5e73c9945b9936ae1eb544e1297 : SUCCESS in 21m 42s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3008cdf7b447411c8a95f6f6327a1695 : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/278983c83ff840dcbbca68751bf3de69 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e7b0b5dff81468c998370eddac4aa6c : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/889c76ad713e442499dab8debb9f24eb : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c0b182a106c4a08a9f8817b2f9913ae : SUCCESS in 58m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c77cba889fe498ea6f3449fe23d1603 : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"87c3b2b5bc6e80c4fd3305b42c8f33172f3f7a1c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 13:19:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66b3ce408a316b87943189bcae8b63bd22822aee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-05 15:33:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1efeb7545ed605e3adfcccdaae6e198046a038c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:35:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"328db992fca3585dcfd8aa55e2e4552da21a19ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:18:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23023748d0842f0980dcfc4991b5801 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e19658ff2f47d29313ae59a73d7518 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4159f109cf5f4a869ca01ff92e5b5eb1 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/847346ca34224d5d8e97e1b70c12a661 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c641162b08d746b48577d3eb1863518b : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f89c2eaebf7464b9884a3d80df1e280 : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ebfc3c64ea694e8191d00a4cf98304f4 : SUCCESS in 11m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/42e0889f9e324a869b61eb77fdae557f : SUCCESS in 17m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/974bd6a3913a486e869dcc550c117be6 : SUCCESS in 10m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a896d813e72449a1a11f42de50a0f074 : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b4664a5b73bd409fa025e0e74a058c74 : SUCCESS in 5m 38s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/584afe16c2ae485fa80def0e5715c26f : SUCCESS in 8m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/eb91e9ca686743a2bdb994d54904c0e1 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2f77cb9bda044df9a967ccd8e6da367 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5db796fbd1c34ef1a489ef8e07f95c65 : SUCCESS in 9m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/922f89ee6fef42f7bc13f1305742b3b7 : SUCCESS in 11m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/14632708446240ce9e65a0ba2412297c : SUCCESS in 6m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/111f31217a27422590290f4d501d6de4 : SUCCESS in 10m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/faaea20aea614d089078b9160a6cdcfe : SUCCESS in 7m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bd543d602f774e77a0746896608b7440 : SUCCESS in 26m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/98bcc23cc9ae4614a55fe573d3219833 : SUCCESS in 14m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ec4e975f4c374730899e50db80272ecb : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f450d276bc34a89b90575a21bd835c0 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d60c4b1bab1445f89d9368a5e857ce5 : SUCCESS in 21m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9e9fc0635f1e4cb7b0207b8d58db28b6 : SUCCESS in 7m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/38fc909d85bc4c63a79884cfd3285cae : SUCCESS in 8m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e4facf1256043699404cbb1b0fd5c56 : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/53ab2e22d3974110858242ce48df561e : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/581b6b2b6d7848388fb03a76f89b9aca : SUCCESS in 53m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0095064436474ffda7339dd052af7b21 : SUCCESS in 1h 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9e96160942998544fbdb1a00783ca2cce94bff35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:18:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db82b7ca080f61b20b462d372ae8935927494c86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:20:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8af8618316f4037b3f245e3afa38767 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8e257c6667c4fc7239e38b6bcebd72e20a89524","revisions":{"d8e257c6667c4fc7239e38b6bcebd72e20a89524":{"kind":"REWORK","_number":1,"created":"2022-05-05 09:57:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/840622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/840622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/840622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/840622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/840622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/840622/1"}}},"commit":{"parents":[{"commit":"3c70957648ac44105a755990bde97847d4225644","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c70957648ac44105a755990bde97847d4225644"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 09:57:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 09:57:23.000000000","tz":0},"subject":"update constraint for pbr to new release 5.9.0","message":"update constraint for pbr to new release 5.9.0\n\nmeta: version: 5.9.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Idba5a8c8714f5b1d0e9f1de1d26ae5db9ae5dced\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib5fde11388340c73da79f8faac9297d2a9851e7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8e257c6667c4fc7239e38b6bcebd72e20a89524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8e257c6667c4fc7239e38b6bcebd72e20a89524"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
