)]}'
{"id":"openstack%2Frequirements~840624","triplet_id":"openstack%2Frequirements~master~Iff7d67c2c4a5c19c73a26d047bd7b631cd0f8edd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iff7d67c2c4a5c19c73a26d047bd7b631cd0f8edd","subject":"update constraint for openstack-doc-tools to new release 3.3.1","status":"MERGED","created":"2022-05-05 09:57:29.000000000","updated":"2022-05-05 17:03:50.000000000","submitted":"2022-05-05 17:02:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840624-new-release","meta_rev_id":"d9043203cf5fcdec5425dc368d431b76285c218e","_number":840624,"virtual_id_number":840624,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 17:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-05 15:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-05 13:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-05 15:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 10:15:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 11:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 13:20:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-05 15:33:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"652ddd393f81a2930c281b779becbb7d8cfb27cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-05 09:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"360e4f3aa0156d9509b7e5ca21917cb56fd7f4e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 10:15:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c0c4dfd79fcc4602b32ddcf405cef485 : FAILURE in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"815f245b9a6af0992ada5056bb5f2709c01c12b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 11:22:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec69461ff4534b759f6b725ab4f02243 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/546b9f630fe146caabe0afdbad072d12 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9cc67cc5aff42d492e87901277775f3 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdbb5d13d0ec4404bc22b22c369b6a31 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b514964d60e47488d18ab6b540b71c9 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df721b777f5a4d45bb0c0cbfa8316320 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59014fe514984f6ea6a0cb475eb8f5a2 : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2507d417c84546d3bfafa484e157698f : SUCCESS in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cbdb716893884928b27fcfd93afab2e1 : SUCCESS in 12m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/63c44eb504f846e4a8981fe72c84748a : SUCCESS in 5m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d7c653a885894282909bb7f4d3cd3bb8 : SUCCESS in 10m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ba8c3f3a66c04259bd9ee2219cd53c69 : SUCCESS in 4m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/86b53b1de98a4eeeae7dde0d0b4bec12 : SUCCESS in 5m 22s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f6a99b71657947cdb79c36ccddba4b1f : SUCCESS in 9m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7ae7826eee5c453f8473d5b1a48ffb3c : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09443734755e46ee8b6e1cb82e959efe : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/573c991cd4bb46f2bc0bfb5cc1c7e225 : SUCCESS in 8m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/95321b179762465191f687af5f0b6db5 : SUCCESS in 12m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a8578e4e87af47d69bbc7a48d0ab395d : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8230854afb224de4812631fb6a972c24 : SUCCESS in 8m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a30014b8b4d34796a1e6fccd96b41f4e : SUCCESS in 6m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ae0cbeeafab74317b5290fd47af19bd4 : SUCCESS in 27m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a24540322fb44bd8be5a43ddece1f3e0 : SUCCESS in 12m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fe85ce803f714a778c8aba886d2e9c89 : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/05bffe9f7c164420944c10bdfa39b3a3 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/172466b18d234468b5e1e154f1d7b7cc : SUCCESS in 18m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/661025e1674d44be9a480807dbe62428 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/94250c3502014416b2c15b5116dc4904 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea728028abec4a339f6c1fd2b4c3b2ba : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fece17c39a5a4f3ca16d8ad2c0791ed1 : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ac6c8d42a0e46c992879384ff68a55d : SUCCESS in 57m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a96b3b3e202b42f8b25538f556ee87cc : SUCCESS in 1h 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"51d3b5dbd2f4ba45dcaeb312372cb64203772999","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 13:20:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"767326e1112e6f4a1fa33aca09f56af94c724c05","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-05 15:33:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89f778eacabc901aa60238c98ac2e2f949a51ce7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:35:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8ac85aba22f83fd42284f7b8aec6668d315e716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:02:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f4d719b98b41f79d865cefeb924487 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc01cad30ed04c2191c949d95a71cdfd : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b3bc5b0da1f4f97a5fef6cb0f37ac8f : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eefaaf1a09149f1b431fe926976b441 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/923faa55af854ac4b4c197c9cbcdf293 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a656acc40714f00b0d0a767793e6820 : SUCCESS in 4m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/97149cc45dde4da393f0f09207b88904 : SUCCESS in 11m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2f39c9737c514e18b1b4bcab196c070e : SUCCESS in 13m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9da477f7e7104ba1a6dfd23230551229 : SUCCESS in 10m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/078eff2c492646f9bb87cee34fa3bc55 : SUCCESS in 4m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/de5056bdb87040f8b4b1f1410007deb7 : SUCCESS in 7m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/433ee8aaee7f4418af11c58621fabb12 : SUCCESS in 9m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e23acb06e44c404182cf7f7363b6ecd5 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a53fe62bb2444f0e997edbc7df877ea6 : SUCCESS in 9m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9c0c47db2c9540beb484ebc8ccd13e7e : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1f8f5d5ba70e4a68a4746a7b6d0e8e78 : SUCCESS in 16m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ad18764618764f319028d4fa5235da27 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4648c816232748cc8b5cb12b44c0e90b : SUCCESS in 11m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ee5e40d76b654508a1c0fafccf3eb9d0 : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ebf25e5eba4f4b38ac0cffab4e055a0b : SUCCESS in 23m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f4ee01963d0f471194353fc9f7c4a65b : SUCCESS in 14m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7b411d8c04f94dc98d55a712eba9d98b : SUCCESS in 7m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4041cad1549a4612b37c7fc020c1ad2f : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef371a3936004066826ffded67c0e1a3 : SUCCESS in 19m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c7a7459514794feb96c86a82b86e4800 : SUCCESS in 7m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d78da37f6cea4635b7bf9a6a3668c12b : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc6d315654384f13b39bef9990d6c1f1 : SUCCESS in 8m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/40b51e34fb3641cf8de77c36e96be936 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c632aab8d307452694400f1160757da7 : SUCCESS in 48m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e25cac575422401cb12795e5b35ab3d4 : SUCCESS in 1h 25m 22s","accounts_in_message":[],"_revision_number":1},{"id":"50b7fca5854cb3d3682e8caf4991783568326e47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:02:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9043203cf5fcdec5425dc368d431b76285c218e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:03:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792be06b9113438b989313bfe66bb8fc : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5725f412ca8e79cba006815c508db297d898d344","revisions":{"5725f412ca8e79cba006815c508db297d898d344":{"kind":"REWORK","_number":1,"created":"2022-05-05 09:57:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/840624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/840624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/840624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/840624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/840624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/840624/1"}}},"commit":{"parents":[{"commit":"3c70957648ac44105a755990bde97847d4225644","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c70957648ac44105a755990bde97847d4225644"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 09:57:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 09:57:28.000000000","tz":0},"subject":"update constraint for openstack-doc-tools to new release 3.3.1","message":"update constraint for openstack-doc-tools to new release 3.3.1\n\nmeta: version: 3.3.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Idba5a8c8714f5b1d0e9f1de1d26ae5db9ae5dced\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iff7d67c2c4a5c19c73a26d047bd7b631cd0f8edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5725f412ca8e79cba006815c508db297d898d344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5725f412ca8e79cba006815c508db297d898d344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
