)]}'
{"id":"openstack%2Frequirements~840628","triplet_id":"openstack%2Frequirements~master~I6f81a6fabbbc489eb532bf62d6f534affa347723","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6f81a6fabbbc489eb532bf62d6f534affa347723","subject":"update constraint for tooz to new release 2.11.1","status":"MERGED","created":"2022-05-05 10:01:12.000000000","updated":"2022-05-05 17:24:13.000000000","submitted":"2022-05-05 17:22:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840628-new-release","meta_rev_id":"843e608b8f14129efdc806973112885e67da6024","_number":840628,"virtual_id_number":840628,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 17:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-05 15:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-05 13:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-05 15:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 10:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 11:52:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 13:19:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-05 15:33:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e2a832c03b1c986ea4101c82ca692cad4fdae5d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-05 10:01:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48825e754a9076f3fd885e1fe9f98937fcf32e7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 10:18:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8557ca01ee304c2788bc6c63b0064c65 : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d1eeb830050148c0a70c66c3fce98e5ea074a6f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 11:52:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6cb12ac3d0d41b483922a4ab190404a : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7be597237e84e5791ee1aa5f9cdb9d4 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/472ecd12be734360a1da5b993c5490f9 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d63c6934a8e4975b0d5290d7f50874f : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff33f60e24a54d9e8e64b35763a4e1f9 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44703dade19e4d5396c0370d97993bba : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc5ccbcd1e7f4f0f815ce239316f962c : SUCCESS in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b1cf86c20e44df9aa913a67a7993790 : SUCCESS in 11m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cb27e4c681b543c9b6e9bc91133270e5 : SUCCESS in 16m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2fdd80ed768b4767ad71e01ea5f558c6 : SUCCESS in 5m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a8c2c977fdbb46e29e8bcbcaa69e67e0 : SUCCESS in 12m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d1dab644b1754f138dc6a14852455447 : SUCCESS in 4m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e6c8d697085046379f160d4c82dfcac4 : SUCCESS in 5m 52s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b3b622c0ec654c198afe5ee6411c7e85 : SUCCESS in 7m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0caa779bbbcf4562bedef846c8bc96e5 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/884a6d0c2d2e45b3b03ad10895b75f53 : SUCCESS in 7m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1915eea7e10a4bca95b7af26e02a0243 : SUCCESS in 7m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1099eedabde34bcf88372d760cc832e8 : SUCCESS in 10m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7aa896eba51147fa994cc78a58be0c41 : SUCCESS in 4m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/55b4ebcd433c472dbcbd59a8b2f073c0 : FAILURE in 17m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ed472395ebc84c26ad28249ec6c4d937 : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/80b4ff92a7aa49d19dd2b0b11c66189a : SUCCESS in 29m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e56f6723e10348589e01e255bb0fa246 : SUCCESS in 18m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/820e494ce8f8486ab64e2a76fbab6f17 : SUCCESS in 8m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1657847136542568e509cdc6c88cfa4 : SUCCESS in 7m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa874c229baa4ba8b9a5371d66b16b91 : SUCCESS in 19m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/edb34a19029e4379b8bc91b01e14ed36 : SUCCESS in 5m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8fc6950be3dc4c08bf9d543c34bb2787 : SUCCESS in 8m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1bc075b96b34bddb1b02a6ccf1dc4f4 : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eacb1401bede4e82ab15dda2c33389c3 : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48f34ee969a049749bb18b709bc9067f : SUCCESS in 56m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/762c09d087c64ba19ba387fe587bb017 : SUCCESS in 1h 43m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5beb5ded43c9f3e89729e39b6038714db7c7908b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 13:19:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b9f0e3a6bce8097e16644faec811ec116137590","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:36:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c953b3aa77d04b838b3da35e4152562c : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2bcb316b83055426faf05c58c3a283744279d612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:06:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff7c7511239b495388381c0896032161 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3414a46c0a24485b5bcc60ddc393c56 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f0dc8a1f02d4b87a21718cbcd0dfd11 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3a293ac25864e839ef57e22e773ea9f : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5046819e403048e28dc9d5b44f86c008 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/487f5862f38f4a1b827313335fc8e6f0 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8bd4f58f012c4207a6e7749b9586e126 : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fa59043f83774ea3aa7b05a1eddeb481 : SUCCESS in 10m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8236404a5ee246ae829e0b02df1f4af9 : SUCCESS in 15m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/693271eb9e384f68995466cc219d47d4 : SUCCESS in 5m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/548613fde57f42c1b5bf1ba3e002c101 : SUCCESS in 11m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7aecf101f9cf44ebb491d2156b995710 : SUCCESS in 5m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a41b2ee3d03941bda04b2660a32284b4 : SUCCESS in 5m 35s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/547fa53355ce48ff86b2ec6f7fa0d204 : SUCCESS in 8m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/93084c7e06a24a1e8d32dba1e14907d4 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79f2c45263cb42f8bc21ec531623067e : SUCCESS in 7m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ee6639550494f88a40fb44a4d850272 : SUCCESS in 8m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cfa28a2065c14e9597b61bb213f689ed : SUCCESS in 11m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/40eb8cd377e343a39268247ac0334dea : SUCCESS in 4m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/63c71e038992408ba9d5008e52e266d8 : SUCCESS in 10m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6bc67a33ca0146158051ad17d6f27787 : SUCCESS in 5m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b9e7f9ef99ca40b58750d55b9b2bd58e : SUCCESS in 25m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1d5429434c0a43e8aceb718674f2961e : SUCCESS in 14m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ea31ddd3a8ba4fd0b571e09fca2af49e : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02edc3b9e1494defa403041d608b9278 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8f2405464dc43a5bbe9cf7d9ae29bfb : SUCCESS in 21m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3470eda670d2455bbbe41757868bc437 : SUCCESS in 6m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/512c2d75cc754b429a5b890fe57c1c87 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f14ac12e8b445aa80c2745701a0184b : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d5027de1b8c4499ea78e608b2736af61 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/510f2d431d154d6a949813b829b58312 : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96a392d0eeae41e0a8b937df4251889f : SUCCESS in 1h 44m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1bcd1ae513d016fc4e42e232902304abb42fe04f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-05 15:33:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a3994b2bcd6de81acc152becd965debbf195880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:35:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6632693fc798446bbe2ef60811fd5f08a19b2a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:22:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe27edb1aa84e3b86fc7b435012f071 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83a2f05f4dce470b9390d5cdcda29462 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1de1671761dd4d48b699594c20d09610 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6885161a03c3474a8624b050b808098d : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da494496523f46ff9aca92d1d77a3aa6 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d082139a93cb4f459bc8ed9799bbd3d6 : SUCCESS in 4m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ddb418842f2940519ffa6c8064ac32bd : SUCCESS in 10m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ed1d72f0af84636befed085813f20ba : SUCCESS in 13m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d8cbed285ad440a397a8c5c15d6f02ba : SUCCESS in 11m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c7b5d7d060644510a65709b4a8524b7b : SUCCESS in 5m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f86d217cafc14459be67d54fa91684b5 : SUCCESS in 5m 54s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ecbb006ce6da41e9a9cce5fe9789ec9b : SUCCESS in 8m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9689e3a84f1f41a98bdc8e7d45649a01 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8e4a081d9b34ffeaebd2bbe618b6558 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e1dd2e3e558f4d00b615a4e299a8a909 : SUCCESS in 8m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c83e7dea30be4bfaa4827e612b8d21d6 : SUCCESS in 11m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/38755d3a26474eb7aaa1d0f6f2d670a9 : SUCCESS in 4m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8ae13a6d989c437ca974acc9a1d53a35 : SUCCESS in 11m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/19b46d2b529545d2a8dacb1112fe74a2 : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/23e729b6b81741c3bf82c8f7ac161ba2 : SUCCESS in 22m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f22c8e4dec534ff48daf4b12694ad274 : SUCCESS in 13m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b7825760f7354354b2deef2055674d05 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f18557befe3247a48310ba8ed2c25c9f : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92fc60f0144d4e18a673c23a6e40a593 : SUCCESS in 19m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/974af915d19c43f78f5fc245a7d1899c : SUCCESS in 5m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/840a37f197c74f71bb2dada9b98ad2bc : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ac2f3b1220c4e3180f8d7cf074b187c : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7bdbf7cdc3d54778b9824b4d9d101cc4 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a1c08bb4b4b419585a2b370f8a947fd : SUCCESS in 56m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f6e773de58246ac891b851c5f9a7a42 : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a535f9a22dc28b6daedb862ad1fcc85e4c5dc37e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:22:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"843e608b8f14129efdc806973112885e67da6024","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:24:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04fca7686424e3faa588514936df158 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecba5368c60fc7b8c2b89cf12bb9f5c1b6cc27d0","revisions":{"ecba5368c60fc7b8c2b89cf12bb9f5c1b6cc27d0":{"kind":"REWORK","_number":1,"created":"2022-05-05 10:01:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/840628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/840628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/840628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/840628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/840628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/840628/1"}}},"commit":{"parents":[{"commit":"3c70957648ac44105a755990bde97847d4225644","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c70957648ac44105a755990bde97847d4225644"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 10:01:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-05 10:01:08.000000000","tz":0},"subject":"update constraint for tooz to new release 2.11.1","message":"update constraint for tooz to new release 2.11.1\n\nmeta: version: 2.11.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Idba5a8c8714f5b1d0e9f1de1d26ae5db9ae5dced\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6f81a6fabbbc489eb532bf62d6f534affa347723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecba5368c60fc7b8c2b89cf12bb9f5c1b6cc27d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecba5368c60fc7b8c2b89cf12bb9f5c1b6cc27d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
