)]}'
{"id":"openstack%2Frequirements~840682","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I5c0a1f032773299cc2a67fc33ef30e83703244c4","project":"openstack/requirements","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-05-05 22:39:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c0a1f032773299cc2a67fc33ef30e83703244c4","subject":"Bump pyroute2 to 0.6.6","status":"MERGED","created":"2022-05-05 14:45:07.000000000","updated":"2022-05-05 22:39:07.000000000","submitted":"2022-05-05 22:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840682","meta_rev_id":"a7314b009521bf2a5415db868781c456cd96edba","_number":840682,"virtual_id_number":840682,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 22:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-05 19:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-05 19:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 14:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 16:56:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 19:55:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"de2a9ededb8270b1b7582013df89e8ea5bfeef46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 14:45:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f462a7adc83899d498214714a099ba6293fcb35","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 14:48:16.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"343a493d725383f9ea8ac62c6be775e9f2779cff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 14:59:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/64921ff82a81468f9a29818b7a4ebbed : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a2b9221d784f3b2e93aa0ee7a0d36d5c147fd09e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-05 15:25:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e30c386ac204285e6f89f36578d139c6ede0749b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:37:43.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/23ecc55f27db402aa6765e0e14d888a1 : FAILURE in 10m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0b6bbc229ac236e87fddd3f813f56035742e2294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 16:56:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2d130f620c45a4a964f74fb0050ef4 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9caf8b06031b47008b0cc8a838700f1a : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af841c0f2f0c4dc6a58bca90651606bb : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2ce60c42e8a4fdbb759f4397d8d04de : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8972b7b1ccf4ca68a6f593ead423b28 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3eb811ce3b8b4709b77245ea13cd1076 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a1fae88d03914506ac5dd5ef5d34d695 : SUCCESS in 6m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01ce2d4b05b74b7f95a46e9fa48829ec : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/13d5156e0e544790b138c1708567265e : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5a927b2aad004c00bb3811c25f33192d : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/865d2c2662a4408f9d8052363517a997 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63f6f9231c6b4b1e8eb779cca0fc0ef3 : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c6aab7e84b8441b3ad8a732bc2bb6507 : SUCCESS in 8m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/943dac642fee429aae7a990ce6c0f5d8 : SUCCESS in 13m 49s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2c8d4500d66042b69fe1ae1b7383bd41 : SUCCESS in 4m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f745f6b7b003455d831e83fe704cf8d1 : SUCCESS in 34m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/92065ed6c52f4c8ebe3af3e61e5c5e41 : SUCCESS in 11m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d10a3a0c32994101bb421b976177d443 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/869c8ae1bd054d32ba4b1e5866bfc66d : SUCCESS in 23m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8c4adac92dea49769edfe71879ddaaa9 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a0e71141d4548ff8adc9b92d79e28ec : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1ed529262cb74801b877d6fe76742c14 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/134f96568fb64d9890cd6e27c6807cd9 : SUCCESS in 57m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bfc8b69014b4143bd5e86c6e2a83552 : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9f01c22dd7a20a135beed39c0c9ffc2b8f9c2b3b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-05 19:55:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0831ccf3b483d25b67d2057d54934d944ca3ca09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 19:56:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6332f8db1ea38ee3955e2eee28f8999ae97b60b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 22:39:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc6bd9cc36e4fc2983ccd575ae9d303 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c63deda9ff4424a8f5f239f788104e : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aeca63dc3cc44a8bb95c2e2ac814b59 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7e3c021f0984176803e6cdba7339449 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9f4dd7209c64199bdc9b9b62c13a997 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/025a35b69cd5494fa4212e7653d2faa7 : SUCCESS in 9m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67d972f499bf4b1f8f4ab4778296b09a : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/441500aef44945529d34dcd606976edc : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/62c73ad2426b49c1a7cdceaaf45d2d6d : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4055d958157748a6befd4f0d065cf45f : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24608caed42349c4a1e6b14f1149a561 : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9768ef7632bc4ee2bdeefff63ccc5e21 : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83fc4bb501324f60beefbd911f3eee76 : SUCCESS in 14m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/33cb1dce73ff4b6f88f213edebeee196 : SUCCESS in 5m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c54c4cc4009d401dbe00d022e6f3fb7a : SUCCESS in 25m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d886fa6a95494d5388e71ad741702dac : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/136e4d72db474ef787aa6124cd577659 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/626762c1774f4c8abeef565f971b7a37 : SUCCESS in 17m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa74ff7db2a54e26ad0eb63ced61a3d0 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a5d332edc1745fdb2ba10f92062c0bd : SUCCESS in 7m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c9843fc40fe94078a6366603d111603d : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eb023013a0b460d81f6581bc8684dd8 : SUCCESS in 51m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fb7c4a66c66477ab441ef3ce6210350 : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a7314b009521bf2a5415db868781c456cd96edba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 22:39:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91579a0acb33f0a5d0e89ae7ff096eb549040e44","revisions":{"2415b73d756a3c0a8d4ed0a15257128a725cbe53":{"kind":"REWORK","_number":1,"created":"2022-05-05 14:45:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/840682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/840682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/840682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/840682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/840682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/840682/1"}}},"commit":{"parents":[{"commit":"fd775ce44b1c1d2a4f9b77857a00979c1d169ee0","subject":"Merge \"update constraint for oslo.utils to new release 4.8.2\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd775ce44b1c1d2a4f9b77857a00979c1d169ee0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-05 14:37:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-05 14:43:51.000000000","tz":120},"subject":"Bump pyroute2 to 0.6.6","message":"Bump pyroute2 to 0.6.6\n\npyroute2 0.6.5 has a bug that was fixed in 0.6.6 [0]. This issue was\nreported in stable/xena by an Octavia user [1].\n\n[0] https://github.com/svinota/pyroute2/issues/871\n[1] https://storyboard.openstack.org/#!/story/2010027\n\nChange-Id: I5c0a1f032773299cc2a67fc33ef30e83703244c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2415b73d756a3c0a8d4ed0a15257128a725cbe53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2415b73d756a3c0a8d4ed0a15257128a725cbe53"}]},"branch":"refs/heads/stable/wallaby"},"91579a0acb33f0a5d0e89ae7ff096eb549040e44":{"kind":"REWORK","_number":2,"created":"2022-05-05 15:25:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/840682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/840682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/840682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/840682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/840682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/840682/2"}}},"commit":{"parents":[{"commit":"fd775ce44b1c1d2a4f9b77857a00979c1d169ee0","subject":"Merge \"update constraint for oslo.utils to new release 4.8.2\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd775ce44b1c1d2a4f9b77857a00979c1d169ee0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-05 14:37:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-05-05 15:23:25.000000000","tz":120},"subject":"Bump pyroute2 to 0.6.6","message":"Bump pyroute2 to 0.6.6\n\npyroute2 0.6.5 has a bug (introduced in 0.5.19) that was fixed in 0.6.6\n[0]. This issue was reported in stable/xena by an Octavia user [1].\n\n[0] https://github.com/svinota/pyroute2/issues/871\n[1] https://storyboard.openstack.org/#!/story/2010027\n\nChange-Id: I5c0a1f032773299cc2a67fc33ef30e83703244c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91579a0acb33f0a5d0e89ae7ff096eb549040e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91579a0acb33f0a5d0e89ae7ff096eb549040e44"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
