)]}'
{"id":"openstack%2Frequirements~840801","triplet_id":"openstack%2Frequirements~master~I4a481c599c1894f38341b417ec0b4c1e01bf2447","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-05-06 03:31:17.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I4a481c599c1894f38341b417ec0b4c1e01bf2447","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-05-06 02:46:04.000000000","updated":"2022-05-06 06:25:33.000000000","submitted":"2022-05-06 06:24:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840801-openstack/requirements/constraints/noclob","meta_rev_id":"8f027401ac91b25befc34922864ac5d5c20eef9b","_number":840801,"virtual_id_number":840801,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 06:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-06 03:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-06 03:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-06 03:28:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-06 03:29:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-05-06 04:52:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00f7d56092652faa0ae5ba20995ec7db72cf0257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-06 02:46:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b38730b942ab214deed9de56a852651f757b17f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 03:28:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aa53c19d40ae4d0f913aa3ce859c1bc4 : FAILURE in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"467f33d64ed4eb0fa1e62ebd614644cff6838771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-06 03:29:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ac7a54b8261b4a57409fd3740465684b7a9ae78","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-06 03:30:03.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"1d1b6a759582f3880f7eb5747342b9afb38d1910","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-06 03:31:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"280658a2f773af43514bd55eeb2fb5ff4c5ef894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 03:49:35.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7b439cd540034f4e9a9559771bc4f242 : FAILURE in 17m 32s","accounts_in_message":[],"_revision_number":2},{"id":"158e5a7454b952a59e365b4db95ef401cf1a195f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:52:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86bd23b057c74edeaed6357e6e525996 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d0d095d1ee4565a10a318c44a1f957 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eda919558b5d4192b61323ae30af7ae9 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeb4aef5fb7c43e7ba7e22072c82a08f : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da6c2d9b514c4ab9a80150706156ecea : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/edbec7b1a9a94debbebc0ab82d25fb1b : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7365b133701d4ae2b0f84bbf4fd3f2dc : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70a237f21a354509867d15f598820df1 : SUCCESS in 11m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3c3ffaa6b0034910aa16a728e332a8ad : SUCCESS in 15m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f74841384f1242a8a764cbe7828efc9d : SUCCESS in 5m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/69b8ecb884ae44a38a12b3007094a755 : SUCCESS in 11m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/68af6a07cb744ab1aea31b5e856ea612 : SUCCESS in 5m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0540d2fdf79d49c9ae03677b8754a64b : SUCCESS in 5m 32s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a9957b70e8124503b97d76b959623770 : SUCCESS in 8m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ac34c314c24e4fd7afaa0041ce338766 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fddcd2feedc4af8a8fa39450cabbce1 : SUCCESS in 10m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9c85f87583744c0a98c383db055ce694 : SUCCESS in 9m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/82a4e6f1806543d4b4fcc8e6502191a6 : SUCCESS in 13m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/55ce706d5c6d4847870aae5583320d80 : SUCCESS in 4m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dd56675223404df892f51ee9b482f8f0 : SUCCESS in 8m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f6ad90b085d4497af4b7a64b357df92 : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f7a9be1d9e2f43d89f12e9f0e098ea66 : SUCCESS in 25m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1a35710619ed42d1892af043bd958a4b : SUCCESS in 14m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ee3dd18e47f74692812fc12e027a0b96 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9f123a0fd6d48039889cc6ac49d3911 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d48730fbee934c8aa35d02f938ad72de : SUCCESS in 21m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/94c0cb2ca4ac49b5a4947b7ef974cdd1 : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4fe164b947ca446e8102bd7e6b57ac15 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a621197202b4c8f955dc857241515b2 : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cce26ea73fbd4ab08ac675886b933ca4 : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7ac1afad5cf4b168977421a747fc736 : SUCCESS in 52m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/195903218173476baace22849722cfad : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":2},{"id":"2ae5216128ed7055ff5a0f5d235f20a15f687b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:53:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e90d9b26324c69be8e1794e34ffacfb2f7d82ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 06:24:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118bf5a4dd51485e96b41e5e8e3a77d5 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cc71f148c584f7daf47db60c4d31ba8 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d51f9fef5b864fb3b88a916060e585de : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c7efd52ded941229c212e47b5508761 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80889b5db9414f7fb55ac9e210407712 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1935539b1b941a794471552183746ea : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dc6575f1b10d4b048921a3bd4e6e9181 : SUCCESS in 14m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/678d2d8017ec475f8625059044bcc1cd : SUCCESS in 14m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/526140f5d419403e8e81a428818ebf4c : SUCCESS in 10m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6b131eb1c79c430c858b1e1e03d90017 : SUCCESS in 4m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fe32bbbf16054c46a3ef309aa5488baf : SUCCESS in 5m 30s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/92dd408d86204c2695c3a3b41cfc67c3 : SUCCESS in 8m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/058fd4499eec4d22b0727ddfa287b854 : SUCCESS in 12m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ff02c8866e1435b85c782831c195eb6 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/abf4dd4cc616459f8e6dae20a54d7553 : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/66740af34c164b97a5a564c5b11e72b7 : SUCCESS in 11m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/82b5f8297046464dace55f10de13ff66 : SUCCESS in 4m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0c14ca6a8eaf4742b23b986b57b180e4 : SUCCESS in 9m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8b10fece83334f17bbe6af2077e93f3f : SUCCESS in 4m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/27a57b1718f347f5848a784eaafd0810 : SUCCESS in 27m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c7bdfec6711347ae8669bf82adf99732 : SUCCESS in 13m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ee8ef60b1301408896cbed4f3739b4e0 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d814002132fb45bab834ffded481bbba : SUCCESS in 6m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22dbe80170704b98b0e1bf00601fb79e : SUCCESS in 18m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5ccbfc3df56d44f8b546fb4179361d93 : SUCCESS in 5m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/da680284701f42c8859060845377654c : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c168e8b5150c463e93b2387eef3b28ee : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/80ae0622c9364323b4455ef5cb831446 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09970050b1d44decb4eeafe2031e6602 : SUCCESS in 56m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7cae1d1daa34410a9d597a57d339262 : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7c8169d1bfc353366a8e4bfadcf3aeb0959ba474","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 06:24:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8f027401ac91b25befc34922864ac5d5c20eef9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 06:25:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f0d0728af84034a38cf41ec84bed85 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fbfcb726e095dc20618d50a9748432d8be8dce40","revisions":{"05a08938dd99731252d26241af1895d260dc0305":{"kind":"REWORK","_number":1,"created":"2022-05-06 02:46:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/840801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/840801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/840801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/840801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/840801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/840801/1"}}},"commit":{"parents":[{"commit":"ea9647140e1df0d79ca796e23a72810302332588","subject":"Merge \"update constraint for oslo.concurrency to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea9647140e1df0d79ca796e23a72810302332588"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-06 02:46:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-06 02:46:00.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4a481c599c1894f38341b417ec0b4c1e01bf2447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05a08938dd99731252d26241af1895d260dc0305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05a08938dd99731252d26241af1895d260dc0305"}]},"branch":"refs/heads/master"},"fbfcb726e095dc20618d50a9748432d8be8dce40":{"kind":"REWORK","_number":2,"created":"2022-05-06 03:29:56.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/01/840801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/840801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/840801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/840801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/840801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/840801/2"}}},"commit":{"parents":[{"commit":"ea9647140e1df0d79ca796e23a72810302332588","subject":"Merge \"update constraint for oslo.concurrency to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea9647140e1df0d79ca796e23a72810302332588"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-06 02:46:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-05-06 03:29:26.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4a481c599c1894f38341b417ec0b4c1e01bf2447\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbfcb726e095dc20618d50a9748432d8be8dce40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbfcb726e095dc20618d50a9748432d8be8dce40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
