)]}'
{"id":"openstack%2Frequirements~841449","triplet_id":"openstack%2Frequirements~stable%2Fxena~I43146b3e48990525f26cd9ed3d1fb1f23b70cfcc","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I43146b3e48990525f26cd9ed3d1fb1f23b70cfcc","subject":"update constraint for sushy to new release 3.12.1","status":"MERGED","created":"2022-05-11 15:53:05.000000000","updated":"2022-05-11 19:17:25.000000000","submitted":"2022-05-11 19:17:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841449-new-release","meta_rev_id":"71fc84fa910bd1c1ccb7629c6f03e96fb086e276","_number":841449,"virtual_id_number":841449,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 19:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-11 17:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-11 17:50:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-11 16:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-11 17:09:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-11 17:50:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f4c5adc23f6bf2b0cac777285c2eafd06aa0151d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-11 15:53:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bf9f5991415078cf9c598d3296b0fe148471e3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:14:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4fd756c74d1144ca80a96c0bba4c408c : FAILURE in 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1a54f0b12c52463e80798e541dabb54253d003f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 17:09:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c27897dd6749a69a1dbcfd80442f15 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a097daebf2a4a4f87a1598e6c8dafc1 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fae2db6f48f748d58c352c1cf33fdbba : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eedf2ab3b82249bda81af31fc683de1d : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37a1726bd6d74faa8f15c204815de834 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d64193a2c4e24fc08c8852d26b017bab : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/03b51317ea3b49d5ae785016c16ce285 : SUCCESS in 12m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/201ecf2d05594559aa4be7e4696c2394 : SUCCESS in 9m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d495946fb9b048fbbd087cb21f9d4ac8 : SUCCESS in 6m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/77e084a7ae73403a94427b117657d1e3 : SUCCESS in 10m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f11279b7742348c6a1d27120d814a1af : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3a20e6cddb4b4c808fba09ae8758da5a : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e3a7271690d402ba80f554588ae580c : SUCCESS in 11m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2a6e99f14cbf44658bf8af11e50e0ae1 : SUCCESS in 9m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/977921fc084a45afa44aefc146877c07 : SUCCESS in 14m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b0dfc5b31ffa4ff0b9216a335a8fc985 : SUCCESS in 9m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/67351eebd3a34924a3518f25e3939e85 : SUCCESS in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/423200e46c804b2581c2cfd48ec029cc : SUCCESS in 24m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/581451198250457db83e7b22f69ad940 : SUCCESS in 13m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/efa33e7341704c688883d295f607ae22 : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d5988209e474cbeae42c53add380ad3 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c28eaf370d34305b44ec815b7e0d72c : SUCCESS in 22m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/52bdf21e42664043939bf33af284ed4d : SUCCESS in 6m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a68e1ca464534fe4b3b89bbcc96b8f36 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21ce5d2822294e389ddbc36949eb875e : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/99b5410108d94133b3b669d53a4e8b84 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3a79b8571934193b08014840004404a : SUCCESS in 52m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32447682afdf4dbc9759668c72a4ed88 : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"03b538f3eceacaeb1995772e5af3cd456e84d43d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-11 17:50:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f890d9e37ea2e56f9fde9f3d9c62cc6f656af5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 17:51:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8beb7bf6557321193cddf055ab65445e8f011b0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 19:17:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2198668b3bd45d480afa323a537ab17 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be87eed2e26a48479f5d530074e6149f : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01f239ae4980437d9ee9eb72351c44f4 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d05f8bc80bd3437a8761421dd416eb87 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f6224545dd44864aad2d02fd5d9b742 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4493afcc0837434284d73cd44deb606c : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e9a08eeebb148c0af809c64057dd71a : SUCCESS in 8m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1a8f77a43c374c3580f574997eb941c9 : SUCCESS in 13m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d386c066b9b542bb9c905300027b63e9 : SUCCESS in 5m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/72abec0927ac4e6c98ba73c3c0239f6b : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66cc70f76244433b9bf38254ad6d55be : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f59f256382674c80a6c3442a2d0e2baf : SUCCESS in 8m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d442f92ba155453db41e1e35f4636bbf : SUCCESS in 20m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7d82ce0e238445d590897a610cfb44de : SUCCESS in 9m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6d8a3eb6a8a74b2f9bbef7a5fa7ab168 : SUCCESS in 4m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/12ed1845be9e4489b863955d88bc26a3 : SUCCESS in 28m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dc2cd00ee9574cae9ed1a9a0df27370d : SUCCESS in 17m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/978394c032294a759b4f48311242b6ab : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0c3620ac3cc4642b3b2292e5031bfb4 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e3182ad1c0f4aafa631ff6506efd2eb : SUCCESS in 25m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b2c4c460c9264487a2a33d9a70814e16 : SUCCESS in 7m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/27ee81def21b463cb2b6ea11200882ec : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e83ea5d96adc42f198f5cf0c4d5f798b : SUCCESS in 8m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7dcb97da773c4b8b8139f15c55fefbe1 : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/492a58a330aa4f988e7a0663a8f0ae08 : SUCCESS in 50m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89dd366e66374bb3aa9ede43fe48dfcb : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":1},{"id":"71fc84fa910bd1c1ccb7629c6f03e96fb086e276","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 19:17:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fe7d4ca3f4ad2f58bcc2bc61ba3fe1d5047d0a4","revisions":{"9fe7d4ca3f4ad2f58bcc2bc61ba3fe1d5047d0a4":{"kind":"REWORK","_number":1,"created":"2022-05-11 15:53:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/841449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/841449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/841449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/841449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/841449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/841449/1"}}},"commit":{"parents":[{"commit":"7ab916154da7bc420fa5a958e3e786b1d9a78945","subject":"Bump pyroute2 to 0.6.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ab916154da7bc420fa5a958e3e786b1d9a78945"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-11 15:53:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-11 15:53:03.000000000","tz":0},"subject":"update constraint for sushy to new release 3.12.1","message":"update constraint for sushy to new release 3.12.1\n\nmeta: version: 3.12.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Commit: Jacob Anders \u003cjanders@redhat.com\u003e\nmeta: release:Change-Id: I7735b2bc66772a1a6bd41f8a03ea15e8b574559d\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I43146b3e48990525f26cd9ed3d1fb1f23b70cfcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fe7d4ca3f4ad2f58bcc2bc61ba3fe1d5047d0a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fe7d4ca3f4ad2f58bcc2bc61ba3fe1d5047d0a4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
