)]}'
{"id":"openstack%2Frequirements~841489","triplet_id":"openstack%2Frequirements~master~I099703bb72291afa9cb7a5130b823334303528d5","project":"openstack/requirements","branch":"master","topic":"bug/1973162","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-05-13 14:33:06.000000000","reason":"Rodolfo Alonso replied on the change"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-05-13 14:18:07.000000000","reason":"removed on reply"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-05-13 17:24:16.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-05-13 17:24:16.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-13 17:24:16.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-13 17:24:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I099703bb72291afa9cb7a5130b823334303528d5","subject":"Revert \"Block neutron-lib 2.21.0\"","status":"MERGED","created":"2022-05-13 10:09:38.000000000","updated":"2022-05-13 17:25:37.000000000","submitted":"2022-05-13 17:24:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"revert_of":841566,"submission_id":"841489-bug/1973162","meta_rev_id":"756fb1174ec045852824b75ace5f25d810172161","_number":841489,"virtual_id_number":841489,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-13 17:24:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-13 14:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-13 14:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-13 10:09:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-05-13 10:09:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-05-13 10:09:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-13 10:09:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-05-13 10:09:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-13 10:37:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"84d667eb78810361ecb9a953427f8bbb8471920e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-13 10:09:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45605a952c6cf597cf3c3d6e4234d67991f1c46b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-13 10:10:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a483c2ae1c2114db674b74ae4bdd953bae8d399f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 10:29:40.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/586d0b80c9af4db6b95dce1aae5fec38 : FAILURE in 16m 33s","accounts_in_message":[],"_revision_number":2},{"id":"35e9ca6b94b2463173d186d85aece1c5ac2237b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 11:51:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fc7ee11e024f3f933d59b0e0bf18aa : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/556efa60cf3442d0a6641789a895ce47 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40234a21e3404c52b019465d45f655e5 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f9bfe87fb6b44f4ad1d98f3ef63aa5d : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca2076723b054cebaf80ffbcd99eae2a : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/093820f03128497488b3b5da42846d01 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a3444b2bdaa4245b86ae6fe4c2aeff0 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0a6802cd33b5447f9aa63353d86850c6 : SUCCESS in 12m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0375517b3b0742a5a2aa7d7898540b88 : SUCCESS in 14m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1f865bad4d2842b6891706770bb253a9 : SUCCESS in 5m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b0414c9171fb4705a5c003bb19c10e70 : SUCCESS in 10m 36s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d3af859e47fe43ffbe5ec3d2db888238 : SUCCESS in 4m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2d05b48626f840739ff8716775337266 : SUCCESS in 6m 04s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/de93417610ff45c484851139904b7d63 : SUCCESS in 9m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05d2fa0c3c6a45608f047b632c583086 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ab44196766b4c0494c4b1a4863f8dc3 : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c8e1d7cdaea74bf68d1dff08cb8f9d6e : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4261a25e44a446558f13787f2112d411 : SUCCESS in 12m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c73d91489d4948bba9108d37412a0363 : SUCCESS in 5m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/81aacd87b6a048338a08770cbbd08c18 : SUCCESS in 9m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dcf9d2f4011545d7a450dae169d4f592 : SUCCESS in 4m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ad80d243eeef4d6f83d594926850e734 : SUCCESS in 24m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/893824d1f5e44ebe963921221adbb83e : SUCCESS in 12m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d2b5877a7aa44144ae4a41105f4b166b : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b72d70f44354907a1b6bc6d5ad77c6b : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f650b33bc95749d1bf3936ea42565d5a : SUCCESS in 23m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/352c6257c7cb4ad48546042bf3a05097 : SUCCESS in 5m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9882fd58c06946b0bf57d0fc49d0aaa2 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f53285e280b49838dd89bd3c47a2401 : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cea6143ee199480d80fe8f02ec0cc92e : TIMED_OUT in 40m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24c9eba3865d41fa9fb37403bbca7fc5 : SUCCESS in 52m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aeade3d2d200447f93c3f60bc9ac0424 : SUCCESS in 1h 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"dc39e6d4345677871f47ec698669ca30ec2953c7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-13 12:27:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"893fd949888a9f9c0979a75a3e716abfba6bd081","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-13 12:42:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28bdc34179543b0b73a4a261d94bf282a2dd400b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-13 12:43:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6b97ad771dd5a5fdfb73f4f19ed27a50e51b210","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 12:59:24.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a1b89b157680499ea2535c7abb46a64d : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":2},{"id":"50dad524b1e43817a9d71df10dce32a99423d1cd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-13 14:18:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"094560cda504f64946b5ced22247766094517ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 14:29:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df064cf9f67b462b9aa1d380854c2b3a : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84186e51e2344a4b30a0e3472b6f396 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51946eaffcce4b448d3947a949b22008 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b92259d0ffe407ca71f847b0e5ba8bc : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/916f3ff006754b0aa63c525e0d08f54c : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f5099c7fb4a4b13b0186783c01acc7f : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9232e5bf6e2545c7aae45cec0617872f : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/71802935ceff42a4a5a38dd1bd6343f4 : SUCCESS in 11m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/57466b643d6549ac83bf1db50ca8ec1a : SUCCESS in 14m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6bb06d3af3614bccb35f1c7e4b27f300 : SUCCESS in 5m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2e8c09b0972043a9959de68d3410e85f : SUCCESS in 10m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1092b25d8ee0494c9d83c98e25bf6700 : SUCCESS in 4m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2d3e562f3d5a402ba8a1360d8620baeb : SUCCESS in 5m 49s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6e86b0dd3b33451781d8b72382520cd2 : SUCCESS in 7m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/22854869759e4c2192e2c716d70e992f : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a9dafb8adcc4311890a6abd6fda271d : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ea1dae0dbf44016b0e0e3447b85c05c : SUCCESS in 8m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6dce9abeb53a467783601b534e98068d : SUCCESS in 15m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/259fdeddf2e84fd093fd1563d8a39a5c : SUCCESS in 4m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ff62ba57f85041f3aa5719f1f8df5dcf : SUCCESS in 9m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d6194fbca55e4b1e958c7e7ab7f32974 : SUCCESS in 4m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9e4d2ce366d34190ad756bd6f14b95f1 : SUCCESS in 24m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7dd7bb1fe52f4e3ab49e1ab3fa61b233 : SUCCESS in 13m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/da01b9a22bca458fac1c9bfeef084955 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c3ba44ec6b39418194cfd779bd9afcc8 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b787717299a8460f8e2f97279ebf6f2d : SUCCESS in 23m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/19f8bee7ad9a448aa87285881162c0b9 : SUCCESS in 5m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3c57b7de1fec4ea0ab53dacc192c8e74 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a230cacd529e4af08b78a1256bdbe390 : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3a5b0e7a7c1c4d61a8bf134d6ec46910 : TIMED_OUT in 41m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d82d74f1a8c243bea237951a5afefa76 : SUCCESS in 56m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/146d700a05ee4763b7308564aaaf32f9 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":2},{"id":"efc2ebf16e23d3b30f79199b714455a2eac245cc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-13 14:33:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"371d87f55b4d2a076aca896e628dbc0ba04e44ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 14:35:01.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb38b4f9d07d47b7aa9b6fe338ecc595 : FAILURE in 14m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3e5b4cf11e9313c2b254422daa578b2d0178bb7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:01:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2e9f9a1dba4ba2871aafba252532b1 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3faa8e0e95d41a1a4f0486f1436552f : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d21cf8e567ff444eae0550eee5c70a93 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99bfa4657c494580be0b4495adbf0533 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/439d524deac34170a12e1f340a527c4f : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3c19cdaec694585b2cce599a08cb294 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4abd82eede54d7aba3a5a6795b7f078 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/532cbbec40334eb0bcd8cd4d91397749 : SUCCESS in 13m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5802fdb94b4140f882db16b44cd791d1 : SUCCESS in 16m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2e7a2986993442d3b1d2a9b9f47275ba : SUCCESS in 5m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7cde91f0e0954b5a89e9a694960aff30 : SUCCESS in 11m 51s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d19c882a5d3041958fb25a8d5efed360 : SUCCESS in 4m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8b98e28012cd4ec19aa212d18cd581d2 : SUCCESS in 5m 12s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2dc38b29d1134c3885391c9b20511e6a : SUCCESS in 7m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/df65e8cd732146c2994be1eb83affe28 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d70a4f938634b4bb61ce23b33218bed : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2832612ca6f44adda9efecaa089c791b : SUCCESS in 10m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f965ff0ec52d4279b41b467c16a31c59 : SUCCESS in 11m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/adb557246780478a9b1204f7c0aea159 : SUCCESS in 4m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/063382938b9d44af9cb8cd984cb44751 : SUCCESS in 8m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/191f9b6929e94654a0227ba494f87dd2 : SUCCESS in 4m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/af445a3ddd8344ac94a4156a7b9cc1f6 : SUCCESS in 24m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cbabb9dc03ec4f938bf7a4df5b6269a3 : SUCCESS in 11m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/82ccad9e0b8a44a98411b19fbbb897ff : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1fc1d4fc524848b4b78b9ce29c7eea6b : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8db689e0357d441c80d7e1c7007f4983 : SUCCESS in 19m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/896fc8422d1d4c2fb952b1a94a3bb154 : SUCCESS in 5m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7888e0e36fb9457b8215251a84b3b75e : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a593764581b346b09ae25fa0855f5c86 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a410506294b4a8ab401ce95ad4116f1 : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e2905b887ba4b188ef2bd657c64252a : SUCCESS in 48m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a6f49be141f4147a1456f4556942e43 : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"5b90c7917ebd04d18bee480901cae58e45de0f01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:03:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49302c102e0a87c63bfe8f3656cc7154222c63fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 17:24:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2cb7055f8a4c878beb4a7b2f1741ac : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8fdd9e4a71847d2860a71a781a69ccb : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a23c4d64c26f4abdbbc8e2e0c152ff3c : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6d527c285d41d6a81e7c96b863e9ab : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe82dd2e3cc44c2e8438d115e2a42cfe : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da37a33eb05e49f28bd18513b8f02028 : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9540dd0ddfcb4f82a1d4ed02856595e4 : SUCCESS in 10m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7aae55182c0041fc9e96b565c7fbc1ba : SUCCESS in 15m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/84b184034ca043579a305af9ffe8709b : SUCCESS in 10m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/cfa0562ff6664cbe83a4678358f0bfb1 : SUCCESS in 5m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fe0f8ca55e714aea843531c7532e1435 : SUCCESS in 5m 59s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/482e518043844339a2e8407fb5b7e1d9 : SUCCESS in 8m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9d80348cfce140558075ee27f360342f : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e568071190ac4801897d7ef996d0f937 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6561eb5cb174c178560ebe5227c719c : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4cf56c4c60ea43f1b1fcf043d0e7b163 : SUCCESS in 11m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/992570aaf9d2427bbc343b35a288d619 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/769046941c4f408f9c32c810de7b46c2 : SUCCESS in 10m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/54a80b5276604e27b656ab4f2fa741de : SUCCESS in 5m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bb0df8e4913e460490dd3f1e9ef1e920 : SUCCESS in 28m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5fd4e966a45649eb8f76b6f91cdd7676 : SUCCESS in 13m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/61a28668eaf9480ebb47e0f3d8374fce : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2e5773cbdfa049018e3edecf1c6009bb : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d44c85d726d41f2935394c3c37c3836 : SUCCESS in 23m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d6b6fb3c7f4249dda841a19a730eae33 : SUCCESS in 6m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/409c2c3e03c94c4e94cd6146b8b3d4df : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a824eee33cbb4c29afd73c0ba9b2a0fa : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3d0f56d8e19f4254940e72288431a7ef : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fe17d7b5b96438ca219320e167d5709 : SUCCESS in 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8552f20877cb4a79a4bd4ce725202bee : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f107c7315c98a454a1ee1edb9820798e99ccd310","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 17:24:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"756fb1174ec045852824b75ace5f25d810172161","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 17:25:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71794767d9574c62a28f7227626ed023 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"488b1727bf79e1d6f1e9f0e086a5531709a7f5ca","revisions":{"29881e0c3ff793b034e20c48eec1a0891bc7f78a":{"kind":"REWORK","_number":1,"created":"2022-05-13 10:09:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/841489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/841489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/841489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/841489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/841489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/841489/1"}}},"commit":{"parents":[{"commit":"5a8e4c5122389d010b74d0d8c84336b3ceb142ed","subject":"Block neutron-lib 2.21.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a8e4c5122389d010b74d0d8c84336b3ceb142ed"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-05-13 10:09:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-05-13 10:09:38.000000000","tz":0},"subject":"Revert \"Block neutron-lib 2.21.0\"","message":"Revert \"Block neutron-lib 2.21.0\"\n\nThis reverts commit 5a8e4c5122389d010b74d0d8c84336b3ceb142ed.\n\nThe error found in LP#1973162 and LP#1942190 seems to be in the Neutron\nL3 code. There is a race condition when two routers try to update a HA\nport. With [1][2], this is more evident as we protect the DB transactions and Neutron always creates two active port bindings for the same port.\n\n[3] is addressing this problem. I\u0027m reverting this patch to test [3]\nwith neutron-lib 2.21.0\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/828738\n[2]https://review.opendev.org/c/openstack/neutron/+/833247\n[3]https://review.opendev.org/c/openstack/neutron/+/841712\n\nReason for revert: \u003cINSERT REASONING HERE\u003e\n\nChange-Id: I099703bb72291afa9cb7a5130b823334303528d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29881e0c3ff793b034e20c48eec1a0891bc7f78a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29881e0c3ff793b034e20c48eec1a0891bc7f78a"}]},"branch":"refs/heads/master"},"488b1727bf79e1d6f1e9f0e086a5531709a7f5ca":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-13 10:10:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/89/841489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/841489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/841489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/841489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/841489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/841489/2"}}},"commit":{"parents":[{"commit":"5a8e4c5122389d010b74d0d8c84336b3ceb142ed","subject":"Block neutron-lib 2.21.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a8e4c5122389d010b74d0d8c84336b3ceb142ed"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-05-13 10:09:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-05-13 10:10:37.000000000","tz":0},"subject":"Revert \"Block neutron-lib 2.21.0\"","message":"Revert \"Block neutron-lib 2.21.0\"\n\nThis reverts commit 5a8e4c5122389d010b74d0d8c84336b3ceb142ed.\n\nReason for revert:\nThe error found in LP#1973162 and LP#1942190 seems to be in the Neutron\nL3 code. There is a race condition when two routers try to update a HA\nport. With [1][2], this is more evident as we protect the DB transactions\nand Neutron always creates two active port bindings for the same port.\n\n[3] is addressing this problem. I\u0027m reverting this patch to test [3]\nwith neutron-lib 2.21.0\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/828738\n[2]https://review.opendev.org/c/openstack/neutron/+/833247\n[3]https://review.opendev.org/c/openstack/neutron/+/841712\n\nChange-Id: I099703bb72291afa9cb7a5130b823334303528d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/488b1727bf79e1d6f1e9f0e086a5531709a7f5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/488b1727bf79e1d6f1e9f0e086a5531709a7f5ca"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
