)]}'
{"id":"openstack%2Frequirements~841570","triplet_id":"openstack%2Frequirements~stable%2Fxena~I666b0ceaf9ca4349890e827be0047ac4df8201f0","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I666b0ceaf9ca4349890e827be0047ac4df8201f0","subject":"update constraint for oslo.service to new release 2.6.2","status":"MERGED","created":"2022-05-12 10:30:46.000000000","updated":"2022-05-12 17:32:40.000000000","submitted":"2022-05-12 17:32:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841570-new-release","meta_rev_id":"14b90bcaefb557390d63c7dd90c996b1850cd9d1","_number":841570,"virtual_id_number":841570,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-12 17:32:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-12 15:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-12 13:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-12 15:20:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 10:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-12 11:57:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-12 13:48:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-12 15:20:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0676a4f241b4dd147015de0ff7b72d4d3a0b9cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-12 10:30:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"920052862f0e75551bc34893d19aa6ae30d3f951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 10:49:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/37e75a136c1a4f56be4504dcd2266d0e : FAILURE in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c8bf10324c2638c97d8d050cc24b3470ec480704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 11:57:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b8480b85c14b66aec68b85691c54e0 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26963b98746445eb9fb0416702d14e98 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dbd9cc2191546519362dc058d6ee8e5 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abec64ebe7eb4369b07fcfc7ee6603bf : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e4f2f421a0c438999d6f704c0e5439e : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f4a66d7539f438c960ea8f2043d2048 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/017b7921704d491b9fce45b77d1e2643 : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/33e467edef3149eeb4a2f134cf0b5040 : SUCCESS in 8m 03s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/63311859ce474443aa176c2d5c6ea1a2 : SUCCESS in 6m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1be9ddc38c724090890e4032bfca315d : SUCCESS in 9m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b73cf88f63df4ec0afd000279d4415d5 : SUCCESS in 5m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/970a6644ba42492897e39aad3a384a1b : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38d378784af94fd79c597b8d594e7323 : SUCCESS in 8m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f4d08ef4c3214d8d9cb749c6710c531f : SUCCESS in 8m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e0ab3999c929491db10d235c235a9dc9 : SUCCESS in 14m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a5e31839ed4a4dd09c4911d8e4f28722 : SUCCESS in 10m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/753f481f76ec4393b98868d3749bf2d4 : SUCCESS in 5m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2d8e991a99dd47e78a6ef0f7fab5d62b : SUCCESS in 22m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/416c240f824d4401a620ada9222cc4dd : SUCCESS in 10m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4bf0a4f289c0476ca35cb0de8362ee09 : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b5c0f55b4e74fe3a57bbcbdd5deb6ec : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0b2624476f6424282bc0f7576c0c6b0 : SUCCESS in 19m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/12bbeb1bd95245268cb91001fca7a139 : SUCCESS in 7m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6ef0040db6184b6da7fbb6f7fe287e65 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/004b4d9d74dd45c699f11e66e6c9f652 : SUCCESS in 7m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90c6c832d871476d86c2d0116d6816af : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a17bcf7f15134d1985a7f603a07375b1 : SUCCESS in 47m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ce75598b2d441488a33555795de550e : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a6075bb90ae5d1bbe8016a9dab547f150003554b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-12 13:48:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0fff53ba8137a2b0ff85a9f6b5861fe75819bf73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-12 15:20:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e449bb764228bd06cce607189a10f5cd799412bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 15:21:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdeaaf631bc52a2fa2bca419fe921b2acac1a217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:32:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc61b71779c2478380cda33a6f55af51 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e06e855065a74abe8103301672a9237a : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af84ccb9415d4f28a77f64738b1d86f7 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e576069fa2f24c4d9d06940ce1e293a4 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e7a417db825422f864b6f3a95b9f42f : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce40fad714004b0dbc92cae5b7dc00ac : SUCCESS in 7m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d7ad6ab1cc294ed2a6fac3a0e1bee224 : SUCCESS in 8m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/35ecd8815a424cd4be7238f973180745 : SUCCESS in 9m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ebb79faf8e0746d2a8e7707e14c72813 : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b71bd9a50812409082e4e37b677199bf : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7fef14bf4f0c4ae4a6831d8561b42e6a : SUCCESS in 10m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/77d2f06af48a4ea690af339bb21843e1 : SUCCESS in 6m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0bec9950e88047f690a134b13a9704c2 : SUCCESS in 12m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a23e2181dce04c2db93e165e9d12bf90 : SUCCESS in 8m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d57a7ffa4b0b4fb09159fd00c139c511 : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/db4f40de8d2645d7a6512cfe660b1d7c : SUCCESS in 27m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1836aa4607d4477ab29b6da053a6e933 : SUCCESS in 11m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/48b2e3ed7b0340ba9996ca1531ade9f3 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6a95c056523f45be9b2dcc146a71079d : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c211c762ff2c497789e757313d461fb1 : SUCCESS in 15m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/54f49be73dd241a2a9ea651df448ce79 : SUCCESS in 5m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8b7cfb5c16d4477bb1b0c4fcb99cfc5a : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c4cfee0211442298240df57877d7f0c : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/83d2b50f372c4dfa97d4c02188ec3388 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/defe598bfc1c4ef4aefc5134a29c2147 : SUCCESS in 46m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cc05951910c476090aae0fd2d28c07b : SUCCESS in 1h 23m 39s","accounts_in_message":[],"_revision_number":1},{"id":"14b90bcaefb557390d63c7dd90c996b1850cd9d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:32:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a7fe5b440f8ba51a47cfdb68ab980fe63556934","revisions":{"1a7fe5b440f8ba51a47cfdb68ab980fe63556934":{"kind":"REWORK","_number":1,"created":"2022-05-12 10:30:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/841570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/841570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/841570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/841570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/841570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/841570/1"}}},"commit":{"parents":[{"commit":"9fe7d4ca3f4ad2f58bcc2bc61ba3fe1d5047d0a4","subject":"update constraint for sushy to new release 3.12.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fe7d4ca3f4ad2f58bcc2bc61ba3fe1d5047d0a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-12 10:30:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-12 10:30:44.000000000","tz":0},"subject":"update constraint for oslo.service to new release 2.6.2","message":"update constraint for oslo.service to new release 2.6.2\n\nmeta: version: 2.6.2\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I9007d7a49437f443b4a68e10b3649415c273afd6\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I666b0ceaf9ca4349890e827be0047ac4df8201f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a7fe5b440f8ba51a47cfdb68ab980fe63556934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a7fe5b440f8ba51a47cfdb68ab980fe63556934"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
