)]}'
{"id":"openstack%2Frequirements~841571","triplet_id":"openstack%2Frequirements~master~Ic8de40c3cffc8f4db53a9666d14699ac3ed801ff","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic8de40c3cffc8f4db53a9666d14699ac3ed801ff","subject":"update constraint for oslo.messaging to new release 12.14.0","status":"MERGED","created":"2022-05-12 10:33:59.000000000","updated":"2022-05-12 22:59:37.000000000","submitted":"2022-05-12 22:58:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841571-new-release","meta_rev_id":"18a474fc45aa649cf896d0b6ab8a7b18df2b94e7","_number":841571,"virtual_id_number":841571,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-12 22:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-12 15:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-12 13:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-12 15:20:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 10:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-12 12:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-12 13:48:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-12 15:20:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8f16b3be52cf9ddfccfcd5ba06d64212a4704b2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-12 10:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8d154f6a65042a286bb6b6694da7f7b09785726","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 10:52:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7d21f5fc6b9a4831a22f738227e398c3 : FAILURE in 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2cdbe298f24f6e1f062553c0a4a8da5e76a3797a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 12:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c738bd75ce14d96864fbff23b06426a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6986e8d3df6a4418a756fc83319d1c2c : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89c3663928254e36850def664a999718 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5db0cd1de18f45fcbdf17fc1e7a54cd8 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bc9fa29822241e39cf1471a6f47d79e : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43c16f53819a42f589bf1f9e43dbe161 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ce1f7f7aa4a445aaafdd8b2440b966c : SUCCESS in 4m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/47f40efc46d2480b96903ff222e6233f : SUCCESS in 11m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e0408bd2264c4f36940d3772c3596d1d : SUCCESS in 16m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d7f60f07a946468bbbcdae1e58ac4145 : SUCCESS in 5m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5099bcf5a9c148d9a8fd3ee321e975bb : SUCCESS in 10m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7d49aa9e41cc4997b08a1f9fd54283c2 : SUCCESS in 5m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/084527efda4c4f368faeb88b9921b5e5 : SUCCESS in 5m 54s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/48afaffc252d4f5ea782bd94d2094f3e : SUCCESS in 8m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/21fc2ee5c10248ee9d4e4ace9abf5472 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/491118a944334110b403574053f9a901 : SUCCESS in 10m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f4ca0e20a64b450d9d1aa8ba08f37b12 : SUCCESS in 9m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b090b6b94c9347c38bd2d9900e799d25 : SUCCESS in 12m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/741d276d85f345dd8fb4beb88fe89c91 : SUCCESS in 7m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/251e78f877394c08a8218e70b80e26b1 : SUCCESS in 11m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8df6b2c98f0544b18037cdfc484fbfc4 : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ab7a5a118338488eb79cb44941028d2e : SUCCESS in 20m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1c54370aebdb499eb8bcd4dfd9cdb671 : SUCCESS in 11m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/813b2e088de74769b10949223cd8f216 : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f9cffd014a2041ce9045602c3fb6205b : SUCCESS in 6m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4cb4c36f2e0b4754900d54ab6f51225f : SUCCESS in 21m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ad6f194928db453092e4cf880b535c87 : SUCCESS in 6m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b4d921a8add14b07accbaf5ad535aa62 : SUCCESS in 7m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b3be633e1ee4fb796993f967225b56f : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ea3cd417bf9140d1bbd04545fc18a8cf : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a241e0b757a4afc8259bb97812f4b8e : SUCCESS in 48m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8bfa751962948f3af9aec46478b83ac : SUCCESS in 1h 31m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a4aec3848f7b9a995572898bb7c4f3d80f6e680d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-12 13:48:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa57b32a039651b168ac9d1d4f5a5e3ac889d45d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-12 15:20:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae1a35bf8d8295594f927cc5e75b0ef90f72e72c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 15:21:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"205942996ae6e7b3f3ea98df1908424bdcbdb875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:32:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bedcbcdc5c6f42bab2524817367f965b : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f46318a0d7440f8a91448881c61e84f : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae28212114b48bdb65ae788d9f47e1f : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1799df3e0a5c4cb1aaa259a52ff3f958 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c49288a8f3cf427689ff7877e02ef44d : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9aca4a203274d17a2e8ae0c5d1681b1 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/24915f70e15b493aa902e5b5af473f5c : SUCCESS in 10m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c9297959ee4449c790cd0b80c62301af : SUCCESS in 13m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/32a00e04f8c645ed8bca97f54f3a59a4 : SUCCESS in 10m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/39332f55125f49468d908c0442a41e72 : SUCCESS in 4m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/67b37ea717294e1797d8ab8df6a42e5f : SUCCESS in 6m 08s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/eb1154d0b79e4e7d882442cadd44ecb6 : SUCCESS in 8m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/55e5fd67e8474b0687f2c842b0b86ca6 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d870a08214c648158b872528ca7a8bc9 : SUCCESS in 9m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3e4f742db2594723b18e865d3abbaf9c : SUCCESS in 8m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/841a298f1ccd46caa21814fff2ac2daa : SUCCESS in 12m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/166307013d5f4b68a0a5bd2e115422de : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/62bacd753e514da0a5c9061b29649ef7 : SUCCESS in 10m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c5e42031fc154037bef66500cf6e4df7 : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1dbb541641444ae799ff34d79453b2a9 : SUCCESS in 21m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d559b8d3be534d23afe45824d92e09ea : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/932da6d8c0814ffab4d84d401d7fe6e6 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a991bebad32a4e7c881e9fc6e49fd41e : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0d2b9974971499ebbbdf8f5fd7d1c97 : SUCCESS in 20m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cbbd425989f246d1b7a2f3b90a045932 : SUCCESS in 6m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/85f6010f6de24c138367d9a6378ea344 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2b1c019b69442ed9f64405d3dc3cb49 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/39f495c119f841589783ccadbae2899e : TIMED_OUT in 41m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c4ea554c18a4d14be44a6e333679a92 : SUCCESS in 44m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df6fc52daef34d539dbfe3e59eb1ca31 : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a41b640e41371387fb2798ccba904c1e46361d1c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-12 18:20:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c0cf4500fcb49f64c8349ae7e2e39cf9f29d259","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:40:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f99948c9f57a47deb5414b74a10ad135 : FAILURE in 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"685e059f3e3838de15a8b04308744b5c30a812d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 19:58:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4a69b928b844e49ed8c56cd0602271 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762fee96ea634165be651be4ef631696 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3528722681364ab9b6611286f30dacb1 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de021425be4d43b4b7b3a2a3e9c8c72b : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9be60c40614841448c579105f1dcd62f : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35b9b47053814cd296478738ad5e9124 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/261cb565bac94b158150137d47efba59 : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8d07c9d63df4bdb9fa1b89bd49390d6 : SUCCESS in 11m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/57d2ce2f09b2490eb7f16b55d21020ce : SUCCESS in 15m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b8ffb4401d5243ffaa5dfd2657c0912f : SUCCESS in 6m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6b5fcf929d9e4a91b5918b1214decb88 : SUCCESS in 10m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/653f75867ca04e73b90f10962d63c185 : SUCCESS in 4m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7bbb83d629df40c0ad36b161f21af5d4 : SUCCESS in 5m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b418cbc3b39d43cda95c0f6d4d67cc33 : SUCCESS in 7m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/800c735daea44fc08e857dbff0af3530 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d19c36aeba64ea29153302727d26848 : SUCCESS in 7m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ad785c9f3eda45d2bd74fdced3d12b45 : SUCCESS in 7m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e49315771caa4e0eae6c2f255bc338c9 : SUCCESS in 11m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6ae46a78d19f4c5981218c775b212ac7 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ce5f88cf9668417ea318c0cff6ec8eb2 : SUCCESS in 9m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ce0dddfc6b184106b697567893cc16b1 : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/27fe043dfa17403c9e52346ec054710c : SUCCESS in 23m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/02321cd7f8a14233aa87f71d30cf1063 : SUCCESS in 11m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0fc86c660da4455d83a626e73fb94072 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55b57879f32746a08aed07a665705e6b : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dcc98d1f94df4a21a1601950b1c47e38 : SUCCESS in 20m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a9f2877e7d944d22ba05185febd5b4c6 : SUCCESS in 5m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8e4cfe9a398e42bd9088a8932a64c3a9 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/adf9a47b05824e6fb0a966001fdebcaf : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1a4586f1e3d549088ecda89c1cae6a43 : TIMED_OUT in 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7afd8fc70cb4064adb6a01c9b9f4699 : SUCCESS in 49m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d43529c813d4a39ae7aa28a816ffc29 : SUCCESS in 1h 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a93806998efc6457bbaca8fa86a074f7c18077f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-12 20:24:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"489bb08d5ae04272832acc991e5ee1e6da5c5f7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 20:41:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c89569b0b3114fb18f69d119def28886 : FAILURE in 14m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1832744e346ff40042322678e15a1ea435b5bdba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 21:35:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a4fc1cc3a94d64974711c8f7dc9413 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86daa33cddb445d5ab0a88498fca530b : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f626b7da4404186b27e5f7b25b97452 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97f3430037ef495da7b4c3440a8aca79 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e65f462cd84d4640840b245d026e9968 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15754d49d7354c2c896e1ae6a10de603 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4f4ab120a334cb2a1a509558c8f8344 : SUCCESS in 3m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e56b3703bb744dbcbad2f76a21bd22cd : SUCCESS in 8m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/014ea43757a847f998d484e5e9bb6040 : SUCCESS in 11m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5edae6329aa34604abc8f24414de2427 : SUCCESS in 5m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/97c684354c194064be648b9bd13f1984 : SUCCESS in 11m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/edb44205522c4314b3481b52bdc7804f : SUCCESS in 4m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9e573ea3817d470cbc455b7a7e79ca10 : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/30a7a0d27ab642a29dbd387193b357cb : SUCCESS in 9m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ec3f2549bda6472195c4fd1afe450ffc : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6487eb806d045789ed05391797b2cf0 : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d0b88da40287472d8e8648a316437a18 : SUCCESS in 8m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a80727b0f9ef48c4ba3acda01c8dcd46 : SUCCESS in 13m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1fda74f29efa494690403af0a5fbd4bd : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/24441721198f4fd18651943e5c5a545b : SUCCESS in 8m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bf8f46a745434c89a88282829591e56d : SUCCESS in 5m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b082920e867e4f168c3e015b58c5e136 : SUCCESS in 26m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/edc51533fc4c44b8a2b84b31062d9625 : SUCCESS in 14m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4eaecb8ba5bf49caa4c6470f4ad26cda : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/07e178671ba7465b83df5844617572c8 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d1f8eb9d7624bbf8b1ea3635f3865bf : SUCCESS in 17m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ad824ecfa9524b04be69cf39efaa2fc1 : SUCCESS in 5m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/48a817e1f6a34a65924dd10d8a4880ed : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fdf1d7f383c4d789905fe3d9ff3bfb1 : SUCCESS in 5m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e630707fab844cb994c2969ea30375b : SUCCESS in 9m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa8fb22abd0a44149474563d8cddfc8a : SUCCESS in 39m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7d641ab1f154c7882c201bc9dedb42f : SUCCESS in 1h 07m 35s","accounts_in_message":[],"_revision_number":1},{"id":"52f033841fbaddf4ca3c17eda52965650b49c4be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 21:36:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72cf5351c375aba74958101755f351665b14cd1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 22:58:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abccfc5062284d00b585883254fe795f : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80532e98a11b4669872640c430b003e5 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cc0a8c9616f41458efe212be8e491da : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49be063a8a5a485bb89eae5c85e4a7ea : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/326e7bc48b2c4b098b5f94efd1678bef : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/224bf4334f3e40d982c96bb1a0122984 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44fafe08d59c4b83ac6042ccb96bce22 : SUCCESS in 10m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae6358cfd1a24c1aacd1c5d3f1aba1a3 : SUCCESS in 10m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7d7a3dac63e94ad187ab474888de4972 : SUCCESS in 9m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d4b4e7cb28f74d49814a2bbb1529897d : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/973d1eab074442429d23a3195305a0f0 : SUCCESS in 5m 22s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d4b159c350494863a1801cf4fdeba5a3 : SUCCESS in 7m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/46be59db65404a508f24730817aaeed3 : SUCCESS in 8m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3bbbd427895e484e9c51b9af5d603c8d : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d76cd10302924155bbebc56ed29f5ab6 : SUCCESS in 7m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6330dfc8d08e4492a36fd7cdf2ab2582 : SUCCESS in 9m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1f40246db99a4f8f9df55c4d286121fd : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b1fa2b7fa65644c898b8da03c708fa4d : SUCCESS in 10m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/487450c365764c998134792af2899d6f : SUCCESS in 5m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6befaa12252c4e0592ace67d23ecf3ea : SUCCESS in 21m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/844807387fc046699f1cd97ada438db7 : SUCCESS in 11m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a8688a0673034df6a109d93df6162c1d : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d88c63d44354926924a460e522930c0 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94c3b28bac5b4871981429480cc26264 : SUCCESS in 25m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6846b1abbc9140d7b8a67ea1ca863909 : SUCCESS in 5m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/862feb8aee664421a6801910b9cbc32d : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac219522fec84c3d9129b635822d695a : SUCCESS in 7m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bf6d334b419f4c138b00318c9d03c41c : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42f3b75e024c4600a7342ebfe4c5a1c0 : SUCCESS in 48m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6c28d025d284bb9a38421ea62e8e6a7 : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":1},{"id":"df2ea69696bf0cb1e1b41669e704a74f0e4d3489","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 22:58:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"18a474fc45aa649cf896d0b6ab8a7b18df2b94e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 22:59:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7907f293bac84374b18710dd01fe687a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"283b6df6adc5945327b5c10fab5f144a5cb1f260","revisions":{"283b6df6adc5945327b5c10fab5f144a5cb1f260":{"kind":"REWORK","_number":1,"created":"2022-05-12 10:33:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/841571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/841571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/841571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/841571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/841571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/841571/1"}}},"commit":{"parents":[{"commit":"02d2e5995fddc1e946908df8a695df78a6c04975","subject":"Merge \"Update upper-constraints for pyScss to 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02d2e5995fddc1e946908df8a695df78a6c04975"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-12 10:33:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-12 10:33:58.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.14.0","message":"update constraint for oslo.messaging to new release 12.14.0\n\nmeta: version: 12.14.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I894137fc85df58bcd71c3b043293b4c62f702ae5\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic8de40c3cffc8f4db53a9666d14699ac3ed801ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/283b6df6adc5945327b5c10fab5f144a5cb1f260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/283b6df6adc5945327b5c10fab5f144a5cb1f260"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
