)]}'
{"id":"openstack%2Frequirements~842306","triplet_id":"openstack%2Frequirements~master~I1431fc8bf6812a450c68945bb611722fd4212f22","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1431fc8bf6812a450c68945bb611722fd4212f22","subject":"update constraint for os-win to new release 5.7.0","status":"MERGED","created":"2022-05-18 10:07:46.000000000","updated":"2022-05-18 19:13:45.000000000","submitted":"2022-05-18 19:11:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842306-new-release","meta_rev_id":"910e7e748b5695f1fde259de67710b07ba0b5cd6","_number":842306,"virtual_id_number":842306,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-18 19:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-18 15:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-18 10:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-18 15:14:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 10:17:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-18 11:09:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-18 11:38:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 15:14:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"777e0e1adb8f99ce1024af5081eef5396e73e1a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-18 10:07:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00d9c673186788a54059fbfa1a83d584489ca65f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-18 10:17:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5ee4e340e9c833765bcbe3b24573555300b76b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:09:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/732cb9783cf04cd687fdf91e79d60d9e : FAILURE in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"458084f479cef9ea8b0a38f296fdd164e40e9107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:38:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730df90322e44694811aff88bd6c9d02 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f033d1852fe74daf91cdbd67a68ab549 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e15cb87c9b5149f2876a768cfbcce849 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b800bcbdb1af49d6b8aa0be1ac4b0009 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/156e6b88cdd145dcbf4920308fa52a60 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33ea06c9af3d456d8e3423c0e14d8f98 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b469d8afab9741ae9e5615ce2eef7a6c : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da876d9e4ead4123b9955d9000ec55fd : SUCCESS in 9m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5ee5a39b515d4c20ba3b7f15d9157969 : SUCCESS in 13m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/43d4bf63b7f74f4e9bd1306541fd8fb7 : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e8085eb4c06246eaa1055927d09a5413 : SUCCESS in 11m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e3166ae54cbd4359bdbddf7be6a92962 : SUCCESS in 5m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9cc950591781469aa86722646df10222 : SUCCESS in 6m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e09f19d2314f4908b344ddbb3a53cc3b : SUCCESS in 7m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a1afa0169db2442aaecebe043b8a339e : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/213fcb1ff88c4bc0b10b20f485bb1732 : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/429db5194c7d4c9f87937e53f998e53e : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/635b228c199245bbb1754fb13c08bfba : SUCCESS in 10m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/393783d941b8456b9696ccd93dd4c8d5 : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cc37358e8fc341e7900c01826a4b4ac4 : SUCCESS in 9m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f5eeb91c6e504966aa7e1c0c6edcca89 : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2b96777a48ec43ac86c4e547907e9b0a : SUCCESS in 21m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9f66c4f7af1145e5ba7070f5f29031bd : SUCCESS in 12m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2e4faa6d10e34aadba99984265f1d3af : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6618a9a58ee4ff4953054831a9db18f : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be279b27ebf042f19057c672a539bb7b : SUCCESS in 18m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bab4689fbb514e1486d1b5b71bdbad5c : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/676b101244c14ccf9816c1165b578d56 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fa04ca764524ce09a8633df17b53ca9 : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/34ce466c42ce4076ab193965d1f0dae1 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/965f30079d114db3aaae545403b98f57 : SUCCESS in 56m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74cf14ce59f84997a41a82bd2aca4e94 : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bba5bef76e33f77c7189f998ed38c6ef94f6cfd1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-18 15:14:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc8e5d72109f02c93c493f3368d5fa4dfe5c3949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 15:17:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9974b24199096ca723f1f7de41e858efa1326025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:11:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6847492e228e4e05b9af4ffe88686c9f : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd5a1919f8614e55a24258bafd890033 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2469a5bc03d6439fa001af90e84ef8bb : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0af004b9b9214c618b610e7b771d8849 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb9905dfa7eb49bf99bb10b16e65ce87 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a1eb1a2e0164b0d97832f7b027aa714 : SUCCESS in 4m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe5de482ab53461bab54d17a89edcec3 : SUCCESS in 11m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cdb203343f8e48f7a546a9baae13ee99 : SUCCESS in 16m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/578574020b5642a190bb5f4335ac86a5 : SUCCESS in 13m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c958c76cfa7c493abce0d9f0f168f31f : SUCCESS in 5m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c47ea92d64204908a1ca3f92f2cefebf : SUCCESS in 5m 57s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/699f9047f2b943a8b0d6d4d4dbcc504e : SUCCESS in 10m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/49db3f9b02764ad8825210ef06d45947 : SUCCESS in 13m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/326ddb21f73d4f6196d55a0a783ac306 : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4dc2870d86e64bfd8bf4c9d543be0d84 : SUCCESS in 10m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/704cfead645e4128bd8c47f22a4db697 : SUCCESS in 13m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/87774111f60741529fda06c650b9c728 : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4b8b6e0a2e644d26bc79db4460f5ff2e : SUCCESS in 10m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d87c6922fc934c1fb7b3e7ed8c8f26b0 : SUCCESS in 5m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/79739ec2c06e43bd879c7cd7ad777450 : SUCCESS in 28m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/58a28b1486bd40aaba5d765a40854655 : SUCCESS in 13m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a55a1a1f3c144f1986a0f1deea0ec68b : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ee2d252ad754508abf6f03d9f3adb88 : SUCCESS in 5m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52179cae7cb644dbbf4cdc8d2e0f7b54 : SUCCESS in 20m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/29f04c1e113947d2afa3927201c68414 : SUCCESS in 6m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7a46daa153254665a89164c85d61fb7a : SUCCESS in 7m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72c709b29e1c4b38bb765ef1b0bb3fd4 : SUCCESS in 8m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/944ed70edf0b4300abadf968a30fec87 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb8e21a741e2434e88fe5b0bd8f4989d : SUCCESS in 51m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e73cf43c7c5e4195988128263571a2d9 : SUCCESS in 1h 39m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3740712f643ba0fe46386f2f7bf836fcdf0a75ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:11:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"910e7e748b5695f1fde259de67710b07ba0b5cd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:13:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d6211502d246148441c521d14721bf : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07e630f04adb728e6684ed398ffe9dc58055e40d","revisions":{"07e630f04adb728e6684ed398ffe9dc58055e40d":{"kind":"REWORK","_number":1,"created":"2022-05-18 10:07:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/842306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/842306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/842306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/842306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/842306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/842306/1"}}},"commit":{"parents":[{"commit":"62e1b9bf0aa84f98d11f759f2e66167aec016aaf","subject":"Merge \"Add prettytable to global-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62e1b9bf0aa84f98d11f759f2e66167aec016aaf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-18 10:07:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-18 10:07:45.000000000","tz":0},"subject":"update constraint for os-win to new release 5.7.0","message":"update constraint for os-win to new release 5.7.0\n\nmeta: version: 5.7.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: Ie9d86d5cbf8f5d64f604a13902ce4150cd8fc557\nmeta: release:Code-Review+1: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1431fc8bf6812a450c68945bb611722fd4212f22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07e630f04adb728e6684ed398ffe9dc58055e40d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07e630f04adb728e6684ed398ffe9dc58055e40d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
