)]}'
{"id":"openstack%2Frequirements~842475","triplet_id":"openstack%2Frequirements~master~I3fa93f27c30af11cfb52b381f2dc4a1fb412ba03","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3fa93f27c30af11cfb52b381f2dc4a1fb412ba03","subject":"update constraint for ovsdbapp to new release 1.16.1","status":"MERGED","created":"2022-05-19 09:44:25.000000000","updated":"2022-05-19 18:17:25.000000000","submitted":"2022-05-19 18:15:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842475-new-release","meta_rev_id":"aa3bedf756ae36b1ca4861e6898f02b3a64eb7d5","_number":842475,"virtual_id_number":842475,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-19 18:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-19 16:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-19 10:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-19 16:44:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 10:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 10:18:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-19 11:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-19 16:44:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4093e6fcc39382197f69f593155ccc34d716ab7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-19 09:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3207a10f7460224fd5d7fdccdf3fe52f7677667e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 10:09:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c014c2d15814b20925d1e20695c9e3e : FAILURE in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ef87141464822898cb3872b31473a425113993e6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-19 10:18:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77b0458fb293d49d68563117d6b3406cbb64e6df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 11:05:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342c382365ad4374a725d9a19082ab47 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b000462ace084a929123882c76fb8630 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac1342935194ad89e1e6c197fced4e2 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a2fcbb4b0444f81bd8ce18d1de548d5 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/379fb6152bf14df0b3ec029042c92450 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eda61edc79f3473bbaabf795ebf8f913 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21561c6f4cba4f16b5bb19f0181a98f8 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/72bb24057bc64b9d824daf2feb21e981 : SUCCESS in 11m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0f5a78600e6a40928579b93105ff7b9d : SUCCESS in 14m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/046316524f19497f95580f6fad690790 : SUCCESS in 5m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2ef11e6c89ad4cec872de8bea756b7bf : SUCCESS in 10m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/82ed3a1b51fa498f9f94563e1188020f : SUCCESS in 3m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d931aa83db224d92ac869124518d6865 : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/30697089f2544cccb0fa8fdd3d247b5c : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/569bcd8e68294f009fe3023982dd7874 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88a2812edd124f049dbce6741706b5c2 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e6be4fd4aeb24de799a79c4ec6ec221a : SUCCESS in 8m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/17fbd4f4e62a475a94952df57e627807 : SUCCESS in 11m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bdaba2c6fa9947b5ac128a66a248e230 : SUCCESS in 4m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a10a94e988c9428abf1b2299ab0dd856 : SUCCESS in 7m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e1fe3db4f065427597a1d8c2ca0f6899 : SUCCESS in 4m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/297c423b1b824ac1b2222aa818d60c8f : SUCCESS in 28m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7a1b86f8c81d4ca088c2d38a266dac46 : SUCCESS in 11m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c63347641f2b4153aee9294837cbe278 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bdc2b9457db54cd89458467986049804 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14ee449927b8487689222cc9af8191d7 : SUCCESS in 20m 42s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c5076289e7904b78b37e188d105efd38 : SUCCESS in 5m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e7b6fcaf723b4ea6968c183480bfed17 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d9d2ec0e41343ada9ccf66a44458792 : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2ec8126a1d404e45a2eeb72e0555eac7 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03a253e9f6f64d96b73fe073c52eaa7e : SUCCESS in 48m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02ad4d4115e34e1c9215160cb50ee7fd : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f78d0e9a80da7ca667cf6a12667c39a67dc01ecb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-19 16:44:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c47e4277dac8ce7462c726321d4a1bf3f250a368","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 16:45:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"366d51162b78175672abeb5bfedc39766983cbd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:15:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793f72af4b944e1abd2d789d56e6b810 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce1d664f64c4e82865522daad2ab7f4 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a1188e24fb6466d96ca2ba062b75f86 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbbb33542178413587681cf738dc8cac : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8bf9e73c28c4ac08ce1ae40d74ef276 : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be474f506b424bb4984352f9cc30f4aa : SUCCESS in 4m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a6719f6e19e41a3ac8bc1cf39c4bbd2 : SUCCESS in 14m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/462f6a4f1bef4c4dad1f59288ecc6335 : SUCCESS in 15m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/22a1bcda920a41a3a402ab685d622f7f : SUCCESS in 13m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/42a8afdc73f0407db0714d32325a8dce : SUCCESS in 5m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cb473a79d00a4f45a3a489a68e36f54d : SUCCESS in 6m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6d29c0bf27c740008be600c63b51b0f7 : SUCCESS in 9m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/000c12e26ded465d90392a903b78db61 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62981032135b480b8e927bb731fbd3f3 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/61b32a42e7d74d658219ee34cb52f708 : SUCCESS in 11m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/84e79e8ecd1046a1b9743bb1b20b4afe : SUCCESS in 13m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ce407d90fcbf4c9599c47fad4f5f3375 : SUCCESS in 6m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ed5ef38ab9fb4e708160f37e58618a1c : SUCCESS in 8m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e6dda42782874ebd8934d5d13bf5ec57 : SUCCESS in 5m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/282603c1b746409a884d1ddc55d4ee6f : SUCCESS in 23m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e1b6674f833e4906b6a68400f83fab20 : SUCCESS in 13m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bdc5815b349f42cea0ea5ecd88bc5282 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f6150f2b1b1406a9736cc350cd05dac : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0da8795ad2b4534b2150cf0a90d15ae : SUCCESS in 24m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d26db17221e24fc09de0bc28d61d42de : SUCCESS in 6m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6bda693bb6f4478ca2670e19061e5d33 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc22d975c50f4f73990e739f0a2c97a4 : SUCCESS in 8m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5568f0eca5cd499ca94462b219895973 : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f24ada242f141c78e1e9d3ebac0a21d : SUCCESS in 48m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bee49eab3f74d6a8376468638426d66 : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"99fff2b30d24fa8908ef75ee6d4a1013ee001868","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:15:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa3bedf756ae36b1ca4861e6898f02b3a64eb7d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:17:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00755805be55447c96cc6d91f9280ae7 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f836e136a746d9cbd4a1bbdf678e95005dbc20bd","revisions":{"f836e136a746d9cbd4a1bbdf678e95005dbc20bd":{"kind":"REWORK","_number":1,"created":"2022-05-19 09:44:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/842475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/842475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/842475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/842475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/842475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/842475/1"}}},"commit":{"parents":[{"commit":"f3ad0aa75f766f92b90c40f1603ea034be60b86e","subject":"Merge \"update constraint for os-win to new release 5.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3ad0aa75f766f92b90c40f1603ea034be60b86e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-19 09:44:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-19 09:44:24.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.16.1","message":"update constraint for ovsdbapp to new release 1.16.1\n\nmeta: version: 1.16.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: elajkat \u003clajos.katona@est.tech\u003e\nmeta: release:Commit: elajkat \u003clajos.katona@est.tech\u003e\nmeta: release:Change-Id: I2a9d7da020d331ad712fa35e0183c7bc45da794d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3fa93f27c30af11cfb52b381f2dc4a1fb412ba03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f836e136a746d9cbd4a1bbdf678e95005dbc20bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f836e136a746d9cbd4a1bbdf678e95005dbc20bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
