)]}'
{"id":"openstack%2Frequirements~842477","triplet_id":"openstack%2Frequirements~master~I4336cb41bad0bc9580cc6bb1fe3a479804c2db61","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4336cb41bad0bc9580cc6bb1fe3a479804c2db61","subject":"update constraint for oslo.log to new release 5.0.0","status":"MERGED","created":"2022-05-19 09:56:05.000000000","updated":"2022-05-19 18:18:37.000000000","submitted":"2022-05-19 18:15:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842477-new-release","meta_rev_id":"c489ff5c8f2710248d805b7121c5c5250893c7f5","_number":842477,"virtual_id_number":842477,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-19 18:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-19 16:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-05-19 10:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-19 16:44:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 10:18:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-19 10:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 11:56:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-19 16:44:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c728d2488d5fea348e60e3dff38efc135ae92a4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-19 09:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"638faf88c4f39216c7a56968a7a8cfccff7d8ede","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-19 10:18:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c2c07e20f7862f4a127743ff3fbafe7dd135d2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 10:29:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/52fa65cb29484076b7fb1e5431348820 : FAILURE in 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"dd06d2a3325fb55d7f87faa9c687cfd83e683775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 11:56:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6785063bd4914181b8bff4526d599246 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73c02c77c614a1197b9448af02a7dff : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d75854d1c444072b273a56dbeb0afef : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fc5672ee8684a3fb94a20f741cb7498 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e11507adec754327a335be9ff9df2291 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66f271892d1344398453d3cde091bcc6 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86168d73b3514725a39bee732f9356f1 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ba8435f0bd1447c8af141f098ab39dd1 : SUCCESS in 10m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6a44e42213354873872fdb29d6e926be : SUCCESS in 12m 46s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0f869425bf214909b08ec423b41a91d7 : SUCCESS in 4m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0c6bce047d2a4645bcce2a118d2b9b8a : SUCCESS in 9m 14s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4edd25ff59e446e5889632f2c20a228a : SUCCESS in 5m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2d64257cd8034c298b596ee6ba60ed8c : SUCCESS in 4m 51s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3a74ed3bb66146bf97f2c8437a2b36e3 : SUCCESS in 7m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d1ef10a96f8344a8bf702dfa163c3e5c : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eec1357f553f430280c535a12d66a7d0 : SUCCESS in 7m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/74b9e19062334db49b9bddc3b7613a08 : SUCCESS in 7m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0b7c5571b6f84245a127bc7504a28684 : SUCCESS in 10m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6487cfdc556f4cf78cd5e6d95a1ce445 : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/597e9eef44c343c996e64a7128a9c88d : SUCCESS in 8m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/23d2395efacd49d99f1541c7c577fc14 : SUCCESS in 4m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/643eb36f32154ed29527bfec1f56edd3 : SUCCESS in 24m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1039918f123a4da5b69b4cf76372d974 : SUCCESS in 11m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6e8dda6342e94163b4f92a8bb5515a1b : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/41bc4bb5f6344f49b25eb402316cee98 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8968a57f9a5a4e3392d93eef8673e1d9 : SUCCESS in 17m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9deba5b657094c1ca067a7a203f272c4 : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/deda0f973872464984b38baee11360ef : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de1df9bde58e475596a004fb15317417 : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0a6f1e1ddbf547e2bb7468047752d893 : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35cb4f7aed0d4bfd98b50cf1a81b36b4 : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/902b2f013812456ca2047c7418bb281a : SUCCESS in 1h 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0101a6dc4e31527fbd914df6a9bad81998cf626d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-19 16:44:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de363b92813d24bb1abf538e69a877203a8a8127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 16:45:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69afc9e37e066195186bdc2de40d28797042f500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:15:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eef0c31d6a742caaa8f78796f540129 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e715b2ac19f14c17a428ec99199f9297 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c323d69d3acd4d1fbf25ef6759d4bb56 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b7758d0a9a84b4fa8f19f4f34c9ca07 : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf7891da1ff84a2a84c6784bca918557 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5ce1210a1e444af988b11b4d32d881e : SUCCESS in 4m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73fa3c52698142aeb26c19a5b2fd15b2 : SUCCESS in 11m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a684746eabe14ed88177bd4c15f7bb40 : SUCCESS in 15m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a6278148e5424c428854cdae9d36843c : SUCCESS in 11m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bfecf7aeef3442c3baceb62ac460d03d : SUCCESS in 7m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/26549a6530224c039643b7e9b953f5ce : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/81caa1d6d7964fee8dc3be7745a306fc : SUCCESS in 9m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/65e99e1e02a64a3eb9a26c7ad30840e9 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8c2b854e3ea48a695b9f875ce821c30 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bbfd883966314a37b3a0072b5b1bde09 : SUCCESS in 9m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1137d11e96e14d98a0d9b9eddab35f83 : SUCCESS in 13m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2d24e5c84f264ed0a2e41cb2493bf1d4 : SUCCESS in 5m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7bedbd55fc124e03b97614e38d2caf95 : SUCCESS in 10m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/92ddad6affdf496eaf96795eb8a505d3 : SUCCESS in 6m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2e51a52b99bf43f4830c224fd6c4a702 : SUCCESS in 22m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/95ed4e169199456f89b9018ab3c5e3c2 : SUCCESS in 14m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dc9fd2dc870643ad9bc6312cbd3654a6 : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1e2e8ba664c4ad0a967160634024ba9 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d47c9a5a4b84d22862d093c4f34e415 : SUCCESS in 21m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3b63135b1af04ea29ff98a366add7973 : SUCCESS in 7m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/09c165b2c6fb422ab6197b598d637232 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f57d7d4ec28b4b8581862a6cb0b3c35e : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/795c2b41a61748b3927821394c02da85 : SUCCESS in 14m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/966c538fdbfc41a78bd6431429a9d90e : SUCCESS in 1h 02m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94c4f6622f604f369ebc629b79adbe39 : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"53ad864f8edea02dff67ecfd32b23fdace107e55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:15:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c489ff5c8f2710248d805b7121c5c5250893c7f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:18:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680422b587034bec90412116fec7d769 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"379f60c6cc12d5223e418a42e9aecd9e1627a992","revisions":{"379f60c6cc12d5223e418a42e9aecd9e1627a992":{"kind":"REWORK","_number":1,"created":"2022-05-19 09:56:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/842477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/842477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/842477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/842477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/842477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/842477/1"}}},"commit":{"parents":[{"commit":"f3ad0aa75f766f92b90c40f1603ea034be60b86e","subject":"Merge \"update constraint for os-win to new release 5.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3ad0aa75f766f92b90c40f1603ea034be60b86e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-19 09:56:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-19 09:56:02.000000000","tz":0},"subject":"update constraint for oslo.log to new release 5.0.0","message":"update constraint for oslo.log to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I2afbd539f4a92b21ff56e95930f30f86aa9e7344\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4336cb41bad0bc9580cc6bb1fe3a479804c2db61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/379f60c6cc12d5223e418a42e9aecd9e1627a992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/379f60c6cc12d5223e418a42e9aecd9e1627a992"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
