)]}'
{"id":"openstack%2Frequirements~842610","triplet_id":"openstack%2Frequirements~master~I23da91a7372262b2773e011238294a99a7a379ab","project":"openstack/requirements","branch":"master","topic":"networkx","attention_set":{},"removed_from_attention_set":{"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-05-22 06:47:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I23da91a7372262b2773e011238294a99a7a379ab","subject":"cap networkx to 2.8 for python3.9 and python3.10 also","status":"ABANDONED","created":"2022-05-19 19:01:20.000000000","updated":"2022-05-22 06:47:35.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"875f57ac51afce7d2537e103894aeb14945070b6","_number":842610,"virtual_id_number":842610,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-19 20:32:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-2,"date":"2022-05-19 19:59:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 19:17:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 19:59:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-05-19 20:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 15:38:42.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"}],"messages":[{"id":"6edfd3b2873ff0c48e74e353b7e1d78c78ad8d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-05-19 19:01:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5503eb46b90c594bf99e69d92b64b860eb681f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:17:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/382f284748da4dcaa687b08814700caf : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4f995f6504bfba5c084abc4f6ad64fba621c3357","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-19 19:59:33.000000000","message":"Patch Set 1: Code-Review-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3d66e318110cdb2559cc5c844e0d06265a4bff98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 20:32:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88fd85384b4d49ecbe5e92537fb19967 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815aee776ab7440fad55f77df8ecd9d5 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51a6f186e0c54790be993fe5737ee595 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a623848ed16e4dee9b715a98b13f4668 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c2c20f8f9aa4c1b9c2487167b1635d6 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/528785f7f9184610a42fdb2ab7135346 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e704803e07c48c7a983d06bfd16f60c : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/baf95c917bab4196981c3ca505460626 : SUCCESS in 10m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7ed6cb2b797b48108740c1074249d708 : SUCCESS in 18m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0aca9b574afb4310a52dc4a39eb91b70 : SUCCESS in 6m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d98f74f6d8784432afcae106711498e5 : SUCCESS in 10m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b568bff10ef94085a14942bd3661e38c : SUCCESS in 4m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5e7534a65e40444c9ec3f50581cfcb6d : SUCCESS in 5m 41s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c6afccbd994d45e7a0f7b17e244e7cf5 : SUCCESS in 7m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a98b5c322aa5414a9eb3da71a3c05847 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e149658a0c34a75a087834265def5c6 : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/79e72847490343a09b9f805034aad2f1 : SUCCESS in 9m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/16426b8ce04b421ca976983213135fbe : SUCCESS in 13m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f5e2187ab0154243b6d4fc020a4ab6cd : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0d52a4ec850941938ec05a138d875407 : SUCCESS in 10m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c724deb27677410da700bad88dca443b : SUCCESS in 4m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/65de85e2b37f4c0e8319f8df1d99544d : SUCCESS in 25m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f46cd5e55de944b0a4d50ad7d37a35fe : SUCCESS in 13m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/807e84e8b8124bd9b09f2bf2afa117a8 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cfcfe87c19d4507a5f9f5305b0d4ad5 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/166f2746dfd94c0d97712cc754e71594 : SUCCESS in 22m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/825355a1cc584ab5a137a806e4a0f773 : SUCCESS in 5m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9c0064d1ea8540769b7179bedfec7ac4 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/748962a2e15b4c6dad04d7a9ac08630c : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7f11e75fff83417cbafcc60a7a31f54b : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fb913394eba410b9cf6856c6b8791dc : SUCCESS in 46m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ef1e5e1a91c41719d6314ed5db2d0a2 : SUCCESS in 1h 28m 28s","accounts_in_message":[],"_revision_number":1},{"id":"77565a20f9981f124294e7495cfe4b67cb5a2212","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-05-20 15:38:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"875f57ac51afce7d2537e103894aeb14945070b6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-05-22 06:47:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac77af5d4fe354189ca4a69d54d3c1ff82b83144","revisions":{"ac77af5d4fe354189ca4a69d54d3c1ff82b83144":{"kind":"REWORK","_number":1,"created":"2022-05-19 19:01:20.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/10/842610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/842610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/842610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/842610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/842610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/842610/1"}}},"commit":{"parents":[{"commit":"f3ad0aa75f766f92b90c40f1603ea034be60b86e","subject":"Merge \"update constraint for os-win to new release 5.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3ad0aa75f766f92b90c40f1603ea034be60b86e"}]}],"author":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-05-19 19:00:58.000000000","tz":180},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-05-19 19:00:58.000000000","tz":180},"subject":"cap networkx to 2.8 for python3.9 and python3.10 also","message":"cap networkx to 2.8 for python3.9 and python3.10 also\n\nIt breaks vitrage\nhttps://161d8e5b05dfdb935360-f7d1006348fe8b0ba5afe04a60cb30e4.ssl.cf2.rackcdn.com/842582/1/check/openstack-tox-py39/6075851/testr_results.html\n\nChange-Id: I23da91a7372262b2773e011238294a99a7a379ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac77af5d4fe354189ca4a69d54d3c1ff82b83144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac77af5d4fe354189ca4a69d54d3c1ff82b83144"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
