)]}'
{"id":"openstack%2Frequirements~842700","triplet_id":"openstack%2Frequirements~master~Icd83f90adf3953023be63baabd8fe5b947a9403e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icd83f90adf3953023be63baabd8fe5b947a9403e","subject":"update constraint for sushy to new release 4.2.0","status":"MERGED","created":"2022-05-20 12:30:39.000000000","updated":"2022-05-20 17:04:38.000000000","submitted":"2022-05-20 17:03:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842700-new-release","meta_rev_id":"54bc8ffe84130dd0393b79961f4f980922666fc1","_number":842700,"virtual_id_number":842700,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-20 17:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-20 15:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-20 15:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 12:47:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-20 14:10:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 15:20:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b87f9b99c4c3b2a384fc3d6c43007532e3f22ebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-20 12:30:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95e655e274af657d07dcfc64022b0de418343fbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 12:47:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/86720e01013543618df18e1ac4f77ed5 : FAILURE in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fc28f49c7d0f1d5ea221b8c6d634f8c1a1a4c4d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 14:10:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0015f681f9c488383e0cf93c42a1c76 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd943147e2c4081a17bffeabc65d66c : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1aaed1c6e9249a9b32ff1ba88e8571f : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e12b674acfd44b1a6d47a3fc756dc07 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07630fa91ae946c38dd8dfe5c5d264c1 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64169c2758d94b36a7762bba567d731d : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd7e2238921a4befb10fb09ba6b98064 : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1229adff6a224cad9b6200e064a09c74 : SUCCESS in 10m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7024467634bc435e9066a9b90c71c9b0 : SUCCESS in 12m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/64ee7ea18c074b10899a15732ee69990 : SUCCESS in 4m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/22d5a5c0528b4c1ea6db3a3f3fed5b8d : SUCCESS in 9m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c3c8094aac8845c0b6578168b0116909 : SUCCESS in 5m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f88473b8cb214a09a113e69dfbe90084 : SUCCESS in 5m 29s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4c4681d1724f4c57b62180776c41e4a3 : SUCCESS in 7m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/30a5968e50e5483280ff3fdb20f4335c : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b22fe1f7e534baf9ec802b9c460e1f2 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/661b586ae5ef4978bf26578977d540b2 : SUCCESS in 7m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/05ce586a4ca04f2e84384721201296f9 : SUCCESS in 11m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/01e0dd3b37df46129b6621d4b4d14352 : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/43060fcd14d74035bc5f1adb68052254 : SUCCESS in 9m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ac25ebcfd0a94b83883ec4345e481d74 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6d7b102c7ae0456ea1253218f68e0177 : SUCCESS in 21m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7ad08fa0b6cc4a218b15e4b5b71e6832 : SUCCESS in 13m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4b35a7f02bcf4e8e88c9eb0ff8ffa3b2 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/67a63939075c452fb5f9d3235c74bf28 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e313f8560d8b4657bd3f77cd0977e4b1 : SUCCESS in 22m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9c9c2e244c7c45f9bfb86ab8315e2008 : SUCCESS in 6m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f23dc2db51b943bfaf9a9d38ed7be9f5 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6050d621e1ef4515be2f2f4e3a9b26ce : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/46d8f30b91f54e38bb2ce506f0b2c8ca : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dc0dd38a14740e0a23c8fd66c712c31 : SUCCESS in 48m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2aed3a1a44343aeb833a55da2b7d8f3 : SUCCESS in 1h 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f8d5dbee45e5a0dc0b9e9167795086c7d28bec1d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-20 15:20:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee1901e936bed022020c63395c832dde344618da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 15:21:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c55798dffe6097fd9a638ee3ca729b41a128749","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 17:03:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b668669fc84d4b9cf7713fba8008fd : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce8d83e98574145a6ecec2ba83431a9 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a48c4e5442b1419bb29fbd5eda6287f2 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f190a095a02f4a28a3590e680db9c453 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47c25d64714941a0b573e6eff0bf16e0 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29c8aa3f21704399bab377c440ccef0c : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9214c2631b644a1ca30fb9cc783b1cfe : SUCCESS in 11m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0bba79630ebc41bfabc8a24fdf411ada : SUCCESS in 14m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e63d09e19c2243c196680e7775a91d24 : SUCCESS in 11m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a4be4e8e6d0d4b818ae621b543abe474 : SUCCESS in 4m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a4028cb177a045db8b4ae0f9c233eac2 : SUCCESS in 5m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5935771a29394e919dea84e4fd6e15e7 : SUCCESS in 9m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/681a2a67439d4727b46de6fa78713c41 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a653399cb0b4fbeabfa011e191fcd07 : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/47790350237e4387a4f4604eca60d3ca : SUCCESS in 8m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c25c42a7f6194b49a2f3c3ab0961175b : SUCCESS in 13m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a3255377497c4eac8c7f6968728650b4 : SUCCESS in 6m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f93b09af3ea94be19e78ce8567c72714 : SUCCESS in 9m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a3ec7c51d01d43fc847a4118ccbd052f : SUCCESS in 7m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3c52653fdaef4b2bb0f72f4917028548 : SUCCESS in 25m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/455469c359db4744bad5023bc1d871e8 : SUCCESS in 12m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/80c4cd08ee984cf29cc2af63f4f7c4a9 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d609beef797f484e8d9edc9f5b13226e : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef879118d6944b1aab20bfaacf3bf5d3 : SUCCESS in 18m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/824e597363934637aca79a361a33fdae : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1a9479f342bc4412874e2d057f9eef4b : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5839b7840a54424281b9a934c3fc5f7f : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e5b2338b076b425f9425477d10dab8d1 : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b70688da937343cf8069b65d1aa4e627 : SUCCESS in 56m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/016af25a63464ce4b8ade4f6a6935229 : SUCCESS in 1h 37m 16s","accounts_in_message":[],"_revision_number":1},{"id":"710d8faa0429477330a3d4b0d11dd6dd373cbff1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 17:03:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"54bc8ffe84130dd0393b79961f4f980922666fc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 17:04:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3704fbb2b1824b8797ab106ce0d1aaf1 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e2a099274a5a1bb50c550a73200d560d7fc8d98","revisions":{"1e2a099274a5a1bb50c550a73200d560d7fc8d98":{"kind":"REWORK","_number":1,"created":"2022-05-20 12:30:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/842700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/842700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/842700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/842700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/842700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/842700/1"}}},"commit":{"parents":[{"commit":"6925696f00d8fbc7034aa8057be09327ffaa1b82","subject":"Merge \"update constraint for os-ken to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6925696f00d8fbc7034aa8057be09327ffaa1b82"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 12:30:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 12:30:38.000000000","tz":0},"subject":"update constraint for sushy to new release 4.2.0","message":"update constraint for sushy to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I78fbf3cc4934a0556403c3f34627931e49b91ffb\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Icd83f90adf3953023be63baabd8fe5b947a9403e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e2a099274a5a1bb50c550a73200d560d7fc8d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e2a099274a5a1bb50c550a73200d560d7fc8d98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
