)]}'
{"id":"openstack%2Frequirements~842801","triplet_id":"openstack%2Frequirements~master~I2ef66c2f8d5c57ed6d2c8f4796f386c5d4b022a9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2ef66c2f8d5c57ed6d2c8f4796f386c5d4b022a9","subject":"update constraint for python-manilaclient to new release 3.4.0","status":"MERGED","created":"2022-05-20 19:41:03.000000000","updated":"2022-05-20 22:35:04.000000000","submitted":"2022-05-20 22:33:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842801-new-release","meta_rev_id":"384e4fac899ece4a33959c61a45097cd7f764844","_number":842801,"virtual_id_number":842801,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-20 22:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-20 21:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-20 21:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 20:03:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-20 20:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 21:09:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d12f9b16fef1860d8eb3d289ddeae50938e6c397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-20 19:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e082f5cd2d1fa3cecda8e3958a6f781800c83711","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 20:03:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c01598f54f4a4ed69cdec0ca843a1c77 : FAILURE in 19m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e4128e3ebcf4c95fe5bf72ee90f7443916ca1bc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 20:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6c59f81bb7484e80d5273feb20aa7d : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eaab95d56014bd4b6fd21a2e21d9fa6 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a9ed7dbdd194cf1827015cf3212d73d : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9979a55647d2471a8e459ccaf5abf633 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6867d267cd0c4436a147022300237225 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5850b6698024054a22f8076c31736fa : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4c346ffc56d44a086aa4900b9449e24 : SUCCESS in 4m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59bb7fd8a4a649f5b2490558c86fb25e : SUCCESS in 10m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1ebe7237836548b79ac60e618c84da4b : SUCCESS in 13m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a0f07c8ac4f547a7ac31ca5b1e5b36d2 : SUCCESS in 5m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/10b5300d497f4ba6b9aa2d5428fabf95 : SUCCESS in 10m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7a60c8a14076498c91f8c3162ecbf488 : SUCCESS in 4m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2d0ba87c2e1b413e8978052e27dd3227 : SUCCESS in 5m 53s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/176bcf1f457543c0ad4a947ed1c6e8b3 : SUCCESS in 7m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cf5f5ea8346d48798d38d043de86e27e : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5af907f69eab40629823b17108991a7a : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a07c3b94da3c40e1917766dca7d0c9d8 : SUCCESS in 8m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ca0b8350da6e46a1bf03d4c7b70c71bf : SUCCESS in 16m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/42934dbda3b840a7802060bcb7238868 : SUCCESS in 4m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5ca4e8aac6384d3fa391b415a0309535 : SUCCESS in 15m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e921f730f95e4ceb89840ef3204326ec : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d0b76e1b694d4972a749e354f68be7bb : SUCCESS in 21m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a1851750951542438a9b0d326fb1daa5 : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5ef81a062d8c4f53bfe97316e72868a7 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aba0e541ffe2447eaa15078301aa8f12 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/156e197370ee42d7af7fbcff094f7567 : SUCCESS in 19m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0d51c10883d144a28f106b100d29af30 : SUCCESS in 4m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fc3caaea09f34bd8ba694fc3294a56e4 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31ca8b7addc642e994e275e03a0c3cc7 : SUCCESS in 5m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2fd75c4238324d1e9577d24522db8abf : SUCCESS in 8m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d799f264338c407aa74223d36dd29e0b : SUCCESS in 41m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca16e0e2a2464a42a7f8a1ac01f2bf23 : SUCCESS in 1h 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cc77d67ad50420dd94899d99e88fb94127d18ead","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-20 21:09:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e10bb7de8f499c56937c51b2c15b0054cfda739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 21:10:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36f03f7a25663a69b29e4cf48c817ae7ee579b64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 22:33:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c6be769e2d47f3b698d069f0850062 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c546c51c15ed4aa09bd4d424aa89f9f6 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94fc81ba811a44ee8473cb7e7d13dba4 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8339ea538ae47948103b9b652bf0468 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25e105bebd4a4e29925e749fbcf97575 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0ccabfea25e49a29e27b0ff834e8b0d : SUCCESS in 4m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/100fd75d281142ac903265930ec9a2c7 : SUCCESS in 8m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/edab2ec6fe1d40b49bb06daf289730ae : SUCCESS in 11m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3ba8ad72cf284f548ebc215e26616e94 : SUCCESS in 10m 53s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d4f77044a40644459ce621475684c395 : SUCCESS in 4m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c14b3fae13814f6eaa53ea9b7ed368bb : SUCCESS in 4m 44s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/833ced96bfb54183899cc2d2fabf9d3d : SUCCESS in 6m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5c70c569d31c465db05838893f91fe59 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46e62934d6ac4092ad3a54e20eac9041 : SUCCESS in 6m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e9ce20514b67493983ae8bbc346c84b6 : SUCCESS in 7m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/598de0ee1e664936acceaa27b18d5e43 : SUCCESS in 10m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ba0d468824a44b4b9a962640f2f0b74a : SUCCESS in 3m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ef67b31ab0494666bf0339037e72cc35 : SUCCESS in 10m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1b5400f83a6c41b596f68a6d3d9760ff : SUCCESS in 4m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/81966b526b67432bb16210a1cbbe6828 : SUCCESS in 18m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1b6663a7902b43b0b67409f6e45b8189 : SUCCESS in 11m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6ec383033ad2490c9c2b6795a7bace02 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cacc5e5b5a8f4ac787034d0f3508156b : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9f8f486d6cd437d93cfe202590aa7a3 : SUCCESS in 18m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f2df940862424e46a3087375f305d174 : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2a7ce2dd580a475f91e96e56c308e44a : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba0442c4b826474cb29a9647f6ea9276 : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1d82200f30214796ae6ff7e56d4a0d5a : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01a947b4d9454939bbce7886d79c7910 : SUCCESS in 47m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/083b3d750f724b9babc2e0c8ab690b74 : SUCCESS in 1h 22m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3f9b321452f14eaa2fae5ff2eaa64f06eda5887c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 22:33:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"384e4fac899ece4a33959c61a45097cd7f764844","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 22:35:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d36ad168c4f411498b31bc852829c32 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e986d35a3ab9cb54750d5fe9deda3323b45e26b","revisions":{"3e986d35a3ab9cb54750d5fe9deda3323b45e26b":{"kind":"REWORK","_number":1,"created":"2022-05-20 19:41:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/842801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/842801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/842801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/842801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/842801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/842801/1"}}},"commit":{"parents":[{"commit":"5cdadd16a18e5a24a4aa0a3c1a584bc57a529cfc","subject":"Merge \"update constraint for keystoneauth1 to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdadd16a18e5a24a4aa0a3c1a584bc57a529cfc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 19:41:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 19:41:02.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 3.4.0","message":"update constraint for python-manilaclient to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: Ifd6a40ae541125f17e140643a9dc9f8e4d6a8895\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2ef66c2f8d5c57ed6d2c8f4796f386c5d4b022a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e986d35a3ab9cb54750d5fe9deda3323b45e26b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e986d35a3ab9cb54750d5fe9deda3323b45e26b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
