)]}'
{"id":"openstack%2Frequirements~842804","triplet_id":"openstack%2Frequirements~master~I046a2af9afe776598c59a5f90048557d83836cbc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I046a2af9afe776598c59a5f90048557d83836cbc","subject":"update constraint for osc-lib to new release 2.6.0","status":"MERGED","created":"2022-05-20 20:10:23.000000000","updated":"2022-05-20 23:43:42.000000000","submitted":"2022-05-20 23:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842804-new-release","meta_rev_id":"f2e2d03d1df90400cde5416dc14999298f283def","_number":842804,"virtual_id_number":842804,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-20 23:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-20 22:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-20 22:01:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 20:27:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-20 21:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 22:01:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"36f5502741d58e95267759124a9569b7f308eb19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-20 20:10:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6202fd97b62f3062d09cadf0d9b90159e5c57f02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 20:27:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/78f1167eb0024c2bb263e60685effff7 : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"cd4f3cde8f51944df4fd0edb7e00985e1f554249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 21:27:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7937a1a8c06f49c39b748c412e72f52a : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f4487641a646f6bbed25b203abdf3a : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e7d43e48e9146eab6119ea7e24dfb66 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/764692668b3b40a8a2c3477dee6e10c6 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eef7a55c6a654116a55380e2e6aeaa0a : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa52b23ae0d1435c80e1f9c8c4d02e5e : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4cc920673f0d4b33b3969aaefb315e64 : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1be239ac8f6e4e5db168c33b0deb6f99 : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6576b8e282cd45b5a12d330729740ea3 : SUCCESS in 11m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/28bf16545e8640f9884e60ce1c693984 : SUCCESS in 5m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/596d6100ca3e426986f2f6e0d741a800 : SUCCESS in 11m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4bdb33eef49e4b8bbcd079f2815a1c78 : SUCCESS in 4m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/09b0401a6a204fddbb411dec02c2e9ef : SUCCESS in 5m 42s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/677a60716a2c4111b09a13dff618ff22 : SUCCESS in 7m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4e46113adb4f4313b3fc98ad315e089f : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b70b3822f0974ea09eb0842501be0614 : SUCCESS in 7m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69f97ea62a644656a6f42c7e3219d84d : SUCCESS in 7m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/369823ecfbae4619bf266ce39f718b25 : SUCCESS in 11m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b8731268e84e41c187ccec7b2673b8db : SUCCESS in 3m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2a03130b9b9740a1adf1d1d6ddb2c09d : SUCCESS in 9m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/55597f51963d4eca97cab95fd5893554 : SUCCESS in 4m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eb4db3998c6c46bc893f0bd7fe04635c : SUCCESS in 19m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d29227d1973247498e352fb51527753c : SUCCESS in 11m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fdf847f1f92046b7ae07ceda45caa7f6 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e491bf7755a5442db03b1adf3ec5c543 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5063730739046b492fd1d105becfd1d : SUCCESS in 17m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cb5fc4f215e7410aa98731f56db3ffbf : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c91f11a496f74b2cac74f201c71651fc : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e68b30d2b8874909b59c6c5fb68349c4 : SUCCESS in 6m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cad826c136274e31bc3e31f2d92f1101 : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15094654759241f7835f8088e75f9d53 : SUCCESS in 38m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/966c402e5f3d4baa8653ea2b3310fe2a : SUCCESS in 1h 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c79c2c1b6480aa64c8d5533f1795089377d6e682","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-20 22:01:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c23a42ddeb5eadac78d02b714ac5d51dde5ef9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 22:03:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f1b6c6f15c96f1f5d3a6c08bf651d74858f868e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 23:41:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40fd459ce6e143828431f7ce6b2d2a60 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1de90891b245a99629356f4bc25dcd : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e49186ed5fc46b890c16696dd16d64a : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59a9479d99204e62b0e0a7971ac07923 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90d8b1e9d4724df198dba25b51e47659 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1bca7347f246497b94e0483fc7b20d4c : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56a99622667647dc89b883a9612119ce : SUCCESS in 11m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/643e58de376e411081b8ca7f6bf70b34 : SUCCESS in 12m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3ff6f2ac89b941a6b6c117900381bc91 : SUCCESS in 10m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ef992b5b81bc413b9e8f702afda64803 : SUCCESS in 3m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d7c5e1534ac644a689b0cd6c045e2a7e : SUCCESS in 4m 18s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d09d069d334045d9b9b493f39288d5a6 : SUCCESS in 6m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b5075e88f3cc49e8899d2dc8bb26848f : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00a65a668c25400987af3af8b86b1693 : SUCCESS in 9m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d364384e7f9240ed959a09206d16b1d0 : SUCCESS in 7m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7361997f576d4798aec0a34f2f37b94d : SUCCESS in 10m 33s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/63659bf393904034a9719f71fcdd7393 : SUCCESS in 4m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ecff8604befd4c75af508d4382ad1df9 : SUCCESS in 9m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bfa1c85126004ccba536323755d715c1 : SUCCESS in 4m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/39a99c7f878c40dcad1f8be9e7cdf5ae : SUCCESS in 22m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f9c3b6261a4446999b26985a84370085 : SUCCESS in 12m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2e08f0ae02974323a493ad3c2c963081 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b727650d704645a3b601180311b4dedc : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd9c7bbc5ed6432bb5d56face16d8ca0 : SUCCESS in 20m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/60db2c3b60434b5485bf7077c93e6ef2 : SUCCESS in 5m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b4d2e47ee3c64846ab308d757a92575e : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3371f0d6281d40518f1c6452692cabae : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1184152dc1974ba5ba7d3c110b439b1e : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9cb0deaab3d4951aba4297a8c2eeb67 : SUCCESS in 40m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/472229a348ea4dca8692d46b98f3af6d : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ea93da0459ce38f9d0ada9a1ea160d0540e58e9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 23:41:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2e2d03d1df90400cde5416dc14999298f283def","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 23:43:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390b4f47420946fabc9f8f943c31a12f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19ca40101184e01d064c8430af884bf717cd0758","revisions":{"19ca40101184e01d064c8430af884bf717cd0758":{"kind":"REWORK","_number":1,"created":"2022-05-20 20:10:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/842804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/842804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/842804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/842804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/842804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/842804/1"}}},"commit":{"parents":[{"commit":"5cdadd16a18e5a24a4aa0a3c1a584bc57a529cfc","subject":"Merge \"update constraint for keystoneauth1 to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdadd16a18e5a24a4aa0a3c1a584bc57a529cfc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 20:10:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 20:10:22.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.6.0","message":"update constraint for osc-lib to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Change-Id: I6d3c4fc821436fcafd33d548986820abe6579f7b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I046a2af9afe776598c59a5f90048557d83836cbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19ca40101184e01d064c8430af884bf717cd0758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19ca40101184e01d064c8430af884bf717cd0758"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
