)]}'
{"id":"openstack%2Frequirements~842809","triplet_id":"openstack%2Frequirements~master~I690ede1797f814ac4999ccb740ca475bc6f66882","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I690ede1797f814ac4999ccb740ca475bc6f66882","subject":"update constraint for python-glanceclient to new release 4.0.0","status":"MERGED","created":"2022-05-20 20:44:41.000000000","updated":"2022-05-21 00:38:39.000000000","submitted":"2022-05-21 00:36:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842809-new-release","meta_rev_id":"496c9ed2f2f4548d7e17ae4eaa6d56beea380f72","_number":842809,"virtual_id_number":842809,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-21 00:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-20 22:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-20 22:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 21:01:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-20 22:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 22:49:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1ed97fa4a8e659e7ee0cc9248cac2e21610ce8d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-20 20:44:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"869a98828e9ca6c95dd65737dc20f11d805394fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 21:01:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/48f63dcaf77246599db5cf527bed5b5f : FAILURE in 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8c10d2e59004d7d7a35fd321a7683934f357fda4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 22:16:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b59f1a30fcee41f39db5bef6bd469dcb : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c0711a932dc4e01a6111aceb94b57e6 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b82f91b08364e348d0bdd555ed140f1 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0596ff0650748408a3f811c1eeaf257 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3897cd2d95146539bcc2c99c99cc676 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e534aac06264b10a8e9e08740b0fa61 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/514afd1bd3344892be23e56216a9bf73 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78a3843d86ec44dd9895ccc83cbd7718 : SUCCESS in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/45f9734916334d72a0e68ce8383b7f80 : SUCCESS in 14m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/70ab5524f925489ea6e95b82872fb4e3 : SUCCESS in 5m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/25e3e6bc49e84954a3f4b597a409eb41 : SUCCESS in 12m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9760ee321d2741b685144c524ef8c83d : SUCCESS in 4m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b5d20ce98a5f4193a6730911cb604366 : SUCCESS in 5m 13s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f675c7b2717f42ccbd039ced02ddbae4 : SUCCESS in 9m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/813e60efc81f4d42877fa97636a76d52 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc67a36ae0eb4a05ac6348ff8e53e3fb : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ee982d09d4e4baba43fbf13f407cd85 : SUCCESS in 8m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/165fe20c7a8a4224bd86ef23d8c6d4c5 : SUCCESS in 13m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3ade6b820de54f3e89e0c488922f78d1 : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/072f1e034a3346e4aee0ccbc35b9ac7c : SUCCESS in 9m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d97367e1b9fa41b993dd6d51a6185978 : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/39712986f84d43409e1e9c6f7f0cc0ea : SUCCESS in 25m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2435c0a8321545c4b0c059eb83fd2321 : SUCCESS in 12m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/89a5da19df984808a95fa7bd16276112 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a11abf804c7649b097fe8d4f1b9cb95c : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79d8e73d744c460d92371698d1b70e1e : SUCCESS in 28m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/70df5c48377e4a90b49c88c53bfea7ce : SUCCESS in 5m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6e6e5a9a4699450b9c02908acc3c4299 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29bbe82a1a2c4b0a96c90cd9a8e4da36 : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a706865cd8f4f0e998bfea92c52e11a : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8e0e7141f054a8d971db21ac347c1c1 : SUCCESS in 52m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d0112a4f4ae4ba5ad6d6edada7c267b : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b686cc9a77ba964ce26f8f340d1566e69e09de08","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-20 22:49:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd85eec2d3cec74d55eeee427fb9e39a4bcb6cd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 22:51:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7108100b4a8a9659a173b71583ca4937bcc6a6a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 00:36:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e6b2b63eff408b979ace5050fbc13d : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19cf0272b8e453b8e9ea1d692c92a72 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2039bb090ac744e1975a21f1cbc43a2d : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17ad7b1d8cb0435ca51f2f6a970c5f33 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e84baa9daad14d33a7370c88a1e2d0ab : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68faa66cde42461793ae04eee264a48e : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc10a0e5a5d845efb8fe2b1c1971ad53 : SUCCESS in 12m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a33e3aab00e44d890a88bced57f2a06 : SUCCESS in 14m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5dd8a79ae2ac4a79bdb21deed90b055f : SUCCESS in 9m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e31e5b2b82294a50bacb7a79cee90894 : SUCCESS in 4m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/40ac9e1d68cf4e20bd3f6d87d23af05c : SUCCESS in 5m 52s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/84369838c7ff469392f68612be6ba76a : SUCCESS in 9m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5f78fd7a88c8458dba2647ae9d08a070 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52df71d2c7ad442c81e1a34fcb08ae79 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca6865653e3943d9879537b4e8f87752 : SUCCESS in 8m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bd7ed4793f654d439e0116fd3e3913f8 : SUCCESS in 11m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/943618c9bfd942248e88933b514414fb : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/31c76795c6e34d58a302fa944c172867 : SUCCESS in 10m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b5f749045650454c9ca7bc1471484783 : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3361e06fc001404d84743d3e41f5605f : SUCCESS in 24m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ba1b24e40df049138ff0f6e0e02fd2b0 : SUCCESS in 13m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/21d83739b77f4542834a2a964582bca2 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d226aeeed904e8f90f87e0443f20dab : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ebfa43bbe8b7435da409cad5a189b2c6 : SUCCESS in 17m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/984f2c557d52409099cf3c21dda08502 : SUCCESS in 5m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/959ea7514d8f45dc91c8ebd50e86f3f4 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab90553948644adc97a16c9ca9d1e9f8 : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/747c3acfb9904da7a96db8bfa8d628bd : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/033990b3fbcc4d8d8da0bf7103724721 : SUCCESS in 37m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd847c8ad0584f849da8c9c267662ac1 : SUCCESS in 1h 30m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d284770f37fab54d24aa2b09dc9c066ee0c4a02c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 00:36:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"496c9ed2f2f4548d7e17ae4eaa6d56beea380f72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 00:38:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c9341e1ac948c98faa503982451cb1 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c864eb1814fc29dbd8c1623776117ad4c346273","revisions":{"8c864eb1814fc29dbd8c1623776117ad4c346273":{"kind":"REWORK","_number":1,"created":"2022-05-20 20:44:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/842809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/842809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/842809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/842809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/842809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/842809/1"}}},"commit":{"parents":[{"commit":"5cdadd16a18e5a24a4aa0a3c1a584bc57a529cfc","subject":"Merge \"update constraint for keystoneauth1 to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdadd16a18e5a24a4aa0a3c1a584bc57a529cfc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 20:44:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-20 20:44:39.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 4.0.0","message":"update constraint for python-glanceclient to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If123bd3ea33e34708cf18f486ae3d9c1aa684a44\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I690ede1797f814ac4999ccb740ca475bc6f66882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c864eb1814fc29dbd8c1623776117ad4c346273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c864eb1814fc29dbd8c1623776117ad4c346273"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
