)]}'
{"id":"openstack%2Frequirements~842820","triplet_id":"openstack%2Frequirements~master~I965d8f9ba8aec996698119c0d56359a94dd46de6","project":"openstack/requirements","branch":"master","topic":"fix-tempest-py36","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-21 14:00:17.000000000","reason":"Ghanshyam Mann replied on the change"}},"hashtags":[],"change_id":"I965d8f9ba8aec996698119c0d56359a94dd46de6","subject":"WIP: tempest py36|7 job failure fix","status":"ABANDONED","created":"2022-05-21 00:31:30.000000000","updated":"2022-05-25 22:24:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee8690a49609b6a4f0dc3dab83e6d4ff2b778342","_number":842820,"virtual_id_number":842820,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-21 04:34:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2022-05-21 14:00:17.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-21 00:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-21 01:20:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-21 03:15:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"e084a47a70843eed62023fc6a38740ec00fd56d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-21 00:31:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc77c7bf85b91ffe5a2951f034de443ffc0c7b93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 00:47:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fde7aa6305ff45dbb539702c71e21a56 : FAILURE in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"376fd596432230daab9f65d3ca82b99dac95eeae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 01:20:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a608c993334be8bb52afbbd9278107 : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd971593b6044172a51f5ef143643c1d : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28a842ffcbb34daa8cfb259ac969fed3 : FAILURE in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22d28c74f08b46d5b7582d621617f0be : FAILURE in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/338d270424bc4f8abd4da2cc2f3f9ed2 : FAILURE in 3m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4a7a9fb6e49459ca8961705363e4bac : FAILURE in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d59259306f714528b9dcfbf2f4bea946 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bdf1f7443c2046079f62d97355a348ab : FAILURE in 4m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e6f57d5edab342f9a3a3031c6fce4e52 : FAILURE in 5m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5cac78028ce14c74820c6f4b092dbd90 : SUCCESS in 5m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/25636237950b4d09a176deb4040e0439 : SUCCESS in 10m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c9f3dae0d40243839fbda99ee530131a : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fa418fa02d9d4c59a64d1237999e3745 : SUCCESS in 5m 31s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/178e3e6f2fbf44a289d71f37fc3104c3 : SUCCESS in 8m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/054dc51a5dcf4f779b791c9a1c34806f : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b76e2bee55a84ffe9658e9818e36fb15 : SUCCESS in 8m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a19ac6f11b23491cb879f8c1e15f4917 : SUCCESS in 8m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2e8e99be2d394bbe928ee2a414035c58 : SUCCESS in 11m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/45466ec5d2cc413aa83c3cce92482ea7 : SUCCESS in 4m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1947996f62e84e268af8a42b7d6b9d6d : SUCCESS in 10m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/313880e117c64bfb9c9183ff87e57220 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/922538560a7242369c7a71f22fe419f6 : SUCCESS in 20m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d02340ab16264260afc873374d7f2508 : SUCCESS in 12m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f732245e04e34f60a75e008fd851330c : SUCCESS in 5m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b023c00257cd40e69a219762a28d5fb1 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d8ebf61858142699c9e3f818bc01737 : SUCCESS in 19m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/faa07c5c572748a989f1bd85d05a3595 : SUCCESS in 7m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2ebc6c65095c411ea226342d5a3f3a2d : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5eb840ecf75e4a9581a919f25d343c4c : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d8501d7d51ee4b168896d38802559e0c : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/714ce23514c743fa924d36f614ae8d09 : SUCCESS in 46m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23323996e94044c9a0b0868ce430e5de : FAILURE in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3b06d81b5c3f6d40b5a6473f42b3b58afee5e7d5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-21 03:15:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"567112ebf0c82087714593981b61e9ed2557a84c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 03:31:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7e89a3b963f14689a4f5558dab2dc345 : FAILURE in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b5c95259a5852ac2cd28b546e8582ad0a445affc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 04:34:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e747d4cff94947acf55355a40967fc : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8cff3d4f6bd4206b4aafdd9e1cbf859 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cdbd21f783a43648d38a0b5ad64257a : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d00a656155b4bc6b49c8a89a5634bb9 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f50f7558a004733916c778a9fbd1738 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20da614e0ff8427180bcf0d142965b8d : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3372ec75fb44ea6bd8341f6f3b9d4a4 : SUCCESS in 4m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0dc79b5c634444abbfbde03704eb876c : SUCCESS in 11m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7bc9ef7fcee94980b6f20431d7d20f39 : SUCCESS in 11m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e958e0a4d7354c1c978509dd6a2a2594 : SUCCESS in 5m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/05ca1c228d9c4a6c8225e7a7e896741c : SUCCESS in 10m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5b46f836361441f68a4be9edbce17914 : SUCCESS in 4m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9629c382cdcd4a12b1d4f04ae143c15e : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2c0ef99cc4f8438ab0cb400438375c35 : SUCCESS in 8m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/35b669f956e649a3861edfe567327efa : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4068d6d069e34ce697c81df0b6b89d8c : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/835e51ac05c548e8957e62b1d8be9f6e : SUCCESS in 9m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/41836cc7fc3d4bedbc01fe815da82c7a : SUCCESS in 10m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/44eca686401349b7ab4e74ca92ecb1e4 : SUCCESS in 4m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/625708fe0992468385f1705c2b576c47 : SUCCESS in 10m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/30b1168542a548d6bb9db894d009a2b4 : SUCCESS in 4m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c389688c824a47e69787717121aca1d5 : SUCCESS in 28m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/933081109bfb49b4ae86760ea71d76e3 : SUCCESS in 13m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/634eafd02f294dd798d2f8fb4b36e4f3 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d33592548a644bba83c57dc08b107569 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee490cfe1a80471eb2455eeb38b11700 : SUCCESS in 19m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d3042329be6641568d248b221120182b : SUCCESS in 5m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/078ed2417d344e21b76fbe8a96d749b6 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32ba09cb07694745b14da03d653fd944 : SUCCESS in 5m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dcbfccc7438a4260a179347236249597 : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f4b312f2236499f99dbf92169fb0d59 : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed7a7531853f49d2b7389f31d36fae14 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f49c404b937ecd54c793d4fc0266b1ae571ede5d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-21 14:00:17.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee8690a49609b6a4f0dc3dab83e6d4ff2b778342","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 22:24:22.000000000","message":"Abandoned\n\ngoing with dropping the py36|7 support from tempest https://review.opendev.org/q/topic:bug%252F1975036","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"606f8b416045237e7b7ca29e57305f0c168609e4","revisions":{"606f8b416045237e7b7ca29e57305f0c168609e4":{"kind":"REWORK","_number":1,"created":"2022-05-21 00:31:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/842820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/842820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/842820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/842820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/842820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/842820/1"}}},"commit":{"parents":[{"commit":"3d3de2a8c146a2140cfa460fccf4b84b171d5cf9","subject":"Merge \"update constraint for os_vif to new release 2.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d3de2a8c146a2140cfa460fccf4b84b171d5cf9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-21 00:30:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-21 00:30:52.000000000","tz":-300},"subject":"WIP: tempest py36|7 job failure fix","message":"WIP: tempest py36|7 job failure fix\n\nChange-Id: I965d8f9ba8aec996698119c0d56359a94dd46de6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/606f8b416045237e7b7ca29e57305f0c168609e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/606f8b416045237e7b7ca29e57305f0c168609e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
