)]}'
{"id":"openstack%2Frequirements~842991","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ib5712399794f57b94ba4b06bbb72c71214854eb3","project":"openstack/requirements","branch":"stable/ussuri","topic":"bump-tempst-pin","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-05-23 20:33:26.000000000","reason":"sean mooney replied on the change"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-05-26 10:53:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib5712399794f57b94ba4b06bbb72c71214854eb3","subject":"bump tempest for stable/ussuri to yoga verion pin.","status":"ABANDONED","created":"2022-05-23 15:46:09.000000000","updated":"2022-05-26 10:53:05.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0dc2b8bb2488e059f91d1283db854f41e4d3c5bf","_number":842991,"virtual_id_number":842991,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-23 18:18:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2022-05-23 20:33:26.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-23 16:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-23 17:32:58.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"a10ee8e82f46644862f42c692661cfa065feddf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-23 15:46:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bdf014f373f78334f17ce762cb305f082b11784","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-23 16:32:17.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"385f55956b5454d293f408b77a2b624a61e13ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 16:47:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e46b417bbf46b088737b50d6d7a3b2 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2abc59fabc70483687179777be1f2a31 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e88ac1831f964a49985c361fa2ff8e22 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b8fffa2fba14232a7ceb3217ce43cec : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc36f4f9071e43cbab75763027d3ce8e : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7116e53a54dd4cc5bfe6ca9bdb2ba70d : FAILURE in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dc90bd2088f94ff8b5ca5d7dadbcaeaa : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/203a07e6c6a34d499df26dccd3fc36dd : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6fc3ca8a7c5c437183894169cd3087da : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48c724cd72a944bd8d849ab3bd120660 : SUCCESS in 7m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6b2383fe13f94fd79648a76d06f1977f : FAILURE in 4m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a83d84759bdf4758a14383ddcf40c29b : SUCCESS in 22m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/23ddaf4f00f64ca88fa274ee3990cac0 : SUCCESS in 12m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d89e8d7c9d304d3cb87d69432ea7265f : SUCCESS in 12m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bec440efc91a46d594593186cd30622a : SUCCESS in 15m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c52a9f48f034547aaaa49b861ce4463 : FAILURE in 3m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/42627fd7caab4b349f403518e411517c : FAILURE in 3m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b9b4f5615a14037b1662c1ebe42605f : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/835f9725a99147769b253531f50762fa : SUCCESS in 9m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d385e02a5184524b4704b6ee1049287 : SUCCESS in 9m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a64a61354a10423e8b8110af6c6096fb : SUCCESS in 49m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7738b6f864ad4d8db85f548ab4dc8900 : FAILURE in 46m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f08c1134a1540bb8e9ae836c5ba60a9 : FAILURE in 15m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/8840f4fbcca84765b7b98319ff1b0e7a : FAILURE in 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"da7b58dc3f701e8faee3a2953c1edbf18d417633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-23 17:01:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"079e4f4bff8a71555d2de8e65f86d3e4f90d52b5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-23 17:32:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b95a77c2c77f6b72ad79654fe74928835efe5f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 18:18:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f4e7e533d04babb84846a8cfa7035e : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09fceaf32ed42cfbe3395393d37b2e3 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca7872d200e64474885833e9d5d6cea9 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8c4c1371b174b85bfcb6ef15db00fcb : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8bf0ed19c6194ba4877c73fc08e3c150 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/096ebe55ceea48788c6e3705c8672693 : FAILURE in 5m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d47839b1a3b94015b636e5cb44bf4c8d : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/751385b0f38f4cf79f86cf4b839687e5 : SUCCESS in 4m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1198902747054e149a0c6d85fc3205d6 : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/31b589902af6466b9aee1518598fa767 : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/189d9b364ac249aeac0658a823a44875 : FAILURE in 4m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/480e5c7da9e94263bf404dde3ff608fb : SUCCESS in 21m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b78a9555af92471cadd0d8412a539a92 : SUCCESS in 15m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/27cee0af857348e3b44638d2d47c4f9d : SUCCESS in 13m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46dc2ad1160b4fbabf1089d1077a6912 : SUCCESS in 17m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/39d9efbacbc141ef8bbde02fd98488da : FAILURE in 3m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d95d749c93ea4ba29c87da73f7a6acf2 : FAILURE in 3m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c406e29790214d41b29dbb2c5da133a3 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/61fe13c295ed4a278c7c222846b61ad2 : SUCCESS in 10m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73bc0adad03a472d9715431a80fe79bf : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c510393f726e446181f883f29ab85f02 : SUCCESS in 53m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9faf4dd4417f474c83cc94112ae84422 : FAILURE in 1h 01m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8862fae658dd460e8a093cdcc4e35918 : FAILURE in 20m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/7a1337bba9cb42b09dc2dbf08f173f8b : FAILURE in 1h 15m 19s","accounts_in_message":[],"_revision_number":2},{"id":"847b9f79f4a8effdf766179043b7420074496920","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-23 20:33:26.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0dc2b8bb2488e059f91d1283db854f41e4d3c5bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-26 10:53:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"daeebdb2c6e277c3e45b688403fd6b49fec6eab5","revisions":{"ab8ef44b509fa30d2de71a52977951576c42ae2f":{"kind":"REWORK","_number":1,"created":"2022-05-23 15:46:09.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/91/842991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/842991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/842991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/842991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/842991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/842991/1"}}},"commit":{"parents":[{"commit":"02ef245c6ee3d16c2982abf4b9e1ad4b929cf340","subject":"Merge \"update constraint for os-net-config to new release 12.3.6\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02ef245c6ee3d16c2982abf4b9e1ad4b929cf340"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-05-23 15:33:51.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-05-23 15:46:08.000000000","tz":60},"subject":"bump tempest for stable/ussuri to yoga verion pin.","message":"bump tempest for stable/ussuri to yoga verion pin.\n\ncurrently we have a number of issue with stable/ussuri\nrelated to issue that are fixed upstream\n\nyoga does not test any of the new default policy so lets\nbump it and see if it work.\n\nChange-Id: Ib5712399794f57b94ba4b06bbb72c71214854eb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab8ef44b509fa30d2de71a52977951576c42ae2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab8ef44b509fa30d2de71a52977951576c42ae2f"}]},"branch":"refs/heads/stable/ussuri"},"daeebdb2c6e277c3e45b688403fd6b49fec6eab5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-23 17:01:28.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/91/842991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/842991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/842991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/842991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/842991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/842991/2"}}},"commit":{"parents":[{"commit":"02ef245c6ee3d16c2982abf4b9e1ad4b929cf340","subject":"Merge \"update constraint for os-net-config to new release 12.3.6\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/02ef245c6ee3d16c2982abf4b9e1ad4b929cf340"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-05-23 15:33:51.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-05-23 17:00:55.000000000","tz":60},"subject":"bump tempest for stable/ussuri to yoga verion pin.","message":"bump tempest for stable/ussuri to yoga verion pin.\n\ncurrently we have a number of issue with stable/ussuri\nrelated to issue that are fixed upstream\n\nyoga does not test any of the new default policy so lets\nbump it and see if it work.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843006\nChange-Id: Ib5712399794f57b94ba4b06bbb72c71214854eb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daeebdb2c6e277c3e45b688403fd6b49fec6eab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/daeebdb2c6e277c3e45b688403fd6b49fec6eab5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
