)]}'
{"id":"openstack%2Frequirements~843722","triplet_id":"openstack%2Frequirements~master~I80d5c5bc2f379a9bac41e4bb40594a253853ee12","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-05-28 19:45:37.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I80d5c5bc2f379a9bac41e4bb40594a253853ee12","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-05-28 13:44:51.000000000","updated":"2022-05-28 20:53:47.000000000","submitted":"2022-05-28 20:52:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843722-openstack/requirements/constraints/noclob","meta_rev_id":"a3da4484e785eac46aaa3f48938c402bca09cc19","_number":843722,"virtual_id_number":843722,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-28 20:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-05-28 19:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-05-28 19:45:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-28 13:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-28 14:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-28 15:36:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"91afe048bb370c0d9df3393609539b9ec9e0bd82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-28 13:44:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd61105ead8212798090c68d2a682b4735b9dc01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 13:55:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2363a9ad8a8d4c06848c1d22209a598f : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0ce66523902c9a89da01ddfacf9a1f7ff1e65a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 14:58:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a781aa7d5c134fdcbe184403c9f2e31f : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4c4b0c3d8f499f82556825070d542b : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20e1424d7fa746fe91eac80d661b0e19 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb28612620bd4938bea565bc05006520 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/172465c68d9549a695f32315e9f43e0a : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a3b318cacd94e2685ec4410a1a5f0e6 : FAILURE in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ead612e1ed2a4d31a0e181b5a0a44070 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0eb4ec75ee7f4197b2764c06ee8e1e62 : FAILURE in 3m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3517a25f5738474c8a0fac50c3787452 : FAILURE in 4m 47s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/29c80f647b234ff8be52de5adbf51490 : SUCCESS in 8m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1b9b33a86a244c78814966f8d5f7eb28 : FAILURE in 4m 54s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e88e6648fab5480a8ec067f18012a94b : SUCCESS in 6m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/efe9b99815524a2d907ded92c6c41fb0 : SUCCESS in 10m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2d8bc40bfe344c149c540cf2cb5abb3e : SUCCESS in 4m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4d95c6862a954975960da92f63312f71 : SUCCESS in 5m 30s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fa2a7a5faca94d2da21c2107204b3ead : FAILURE in 7m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0e870871386841868649357a16361f6e : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d89c56da822b4e1e851d675c32b0ebae : FAILURE in 7m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c64c55ea020247a68363701cec93993f : SUCCESS in 9m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/248d590a2dbf4052ae756d3a650746a7 : SUCCESS in 12m 33s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cccefbdbf2b64de7a48b3413025d5db1 : SUCCESS in 4m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/059941c5b04848db971d8db9bef191b1 : SUCCESS in 9m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cdccb47817d445f681c367c0cab707f6 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bf60b4883c3040a7a03c87c1225b20b5 : SUCCESS in 20m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b6ece3a685ec496796a9178658374c21 : SUCCESS in 11m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/de1acc9ec14c4bd0aca8b85b94d9f3b6 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6e9dc38080314557bd6516573fbe5f4b : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44c006f850c54cac88139f06bb67d927 : SUCCESS in 17m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f17f05dc85ad4e10bea6c243d67db8f2 : SUCCESS in 5m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e2a0317d4af546d887fe5927f9556630 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/16a075c7cd584d728373e631c32dc983 : SUCCESS in 6m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0990a7466a12412da17e9ace3b86742a : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47367df03f7a4401887748bab626e064 : SUCCESS in 44m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d2a9f675b2843119c83b54ea26f8f97 : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e95c1af8c84b340e4f089779f79508923ccf4d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-28 15:36:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e976d86e8d1b9719e1bde2b639d27cac413e176","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-28 15:36:42.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"dab8af35617a18983cd6d0ded53237c5adc3b825","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 15:49:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/605175ddea2e4a1a9b2bacb8a6d4d48b : FAILURE in 8m 50s","accounts_in_message":[],"_revision_number":2},{"id":"78dfd1a8ec1a1d3ca486e366db6be345dcbca9ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 17:10:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba7b3b6ac844cc7a8ecf76ef53676ff : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a90bcad10cde41988db5a0c3bf7372a6 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86cea94af3ae44b482f154c3aed55718 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d23311dde0754fc3a240e9a0d4b82b7e : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10eabb5258554d85a2e7d65ac7dabf8f : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/766b90f1776a4593a038050d33724b91 : FAILURE in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9778c99546ee4cccad942d58a389f302 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee33ba5e93ca4fe78fa8dc2f6829ba59 : SUCCESS in 8m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/40627ba615c5492d9fe0ff673e1c5525 : SUCCESS in 11m 11s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a8fc636aa51b45ed935b689a0fbbb58d : SUCCESS in 8m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ec5c9adb87874bd9b149b747dc7f263b : SUCCESS in 4m 35s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d32331bca34f4a3bbf89001490f0aa04 : SUCCESS in 5m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/657ea0d3e2844880a796090a40536e9d : SUCCESS in 9m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5b47b6ef225f4d24805a248a10bd3159 : SUCCESS in 4m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b7aefaf866024c20835b76c2dec4a19d : SUCCESS in 5m 09s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3cc88eb1c5d6470aab69035516183a90 : SUCCESS in 8m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d7f2b0cf026046afa8192be09d819f06 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fad7cc83454b4afbb451be1fdd9a4d9a : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/43a1e08c32354b388b0396d158204b09 : SUCCESS in 7m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/751425e9c31e4d44aab7a8a37403a92a : SUCCESS in 13m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8de3e0fcba5247c684ff1d5151edb940 : SUCCESS in 5m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/344d4e3cf5424d18b384a6f93f5e4afd : SUCCESS in 7m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c4a38711ac524be6b9b28c10157035e5 : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3de95f5fe7f74b908f9e14869c1b4150 : SUCCESS in 23m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0269d705dfbc4e78b7bdf8c7d6f6400d : SUCCESS in 14m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e5aea9edf7224370b7e72f8bb9f610d8 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/855b15e09c3d475bb297f11af7f04b87 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97fc4e73ef304cc89ea15097c4552d00 : SUCCESS in 21m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/073f88e2054e465980ae0a0b2ac34060 : SUCCESS in 5m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6db18883eeaf450db3a3b2ebd95c1d2d : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22f66a33a4414565aa65de1cdf115e59 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/da3a3965747f40b4929e5e95d67953f3 : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96f9a32690cc456a9dd4d679c4d02c2b : SUCCESS in 49m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fea30aaceec47cbbcc938eb12e50b17 : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7837a0bffb7860ca4008841668273499ee7efb52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-28 18:01:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e016fc8f79438c6d2472d3c6731fd5f6d252ddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 18:15:19.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/358e98baf00746ddbb0b3afcaa4f25d0 : FAILURE in 9m 12s","accounts_in_message":[],"_revision_number":3},{"id":"94c7c917fdda73810bf1f0be48ae29626adbba4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 19:40:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53dbc92396cf4c75b71ce97cf9420ed4 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4684130e1254469b4efd1ef84bd0271 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d7033ae87f04c8a9e7c107c71b24aed : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c782ae960dc84ba1bdeab2973477ec09 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b8e74ae2d05417da13ac3d1ae3891eb : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b323e78cc2944e0c9708494805f22a4f : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fbea08bc874444d6bac864f4c6e1bb7a : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/434e0c75548c4fb58da444b2fbc2c504 : SUCCESS in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/30e75725b96649f3a325be853f4c6cfd : SUCCESS in 10m 38s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/b2fe533ab5a5437c8563e11004739804 : SUCCESS in 7m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6e96eca60c6a451b85ce3677bdcc45ac : SUCCESS in 4m 49s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/481991ecc0b6468ead0d55b591a7faa7 : SUCCESS in 5m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/80875359f632409293c9bc4e772b55d5 : SUCCESS in 9m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bad2e7e8bc87443aa2e5676373b56257 : SUCCESS in 4m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ea8ad57c03214d909622a11d57b87383 : SUCCESS in 5m 15s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/606169561da54132a89834754b8786bf : SUCCESS in 7m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d1f12205810f413493935544ae3e4f96 : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ddf1eda05a24d8db8ba324539e512f1 : SUCCESS in 8m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4e2d607034f544788e121f2e4a05e2cd : SUCCESS in 6m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8831404d3b90406f905e567c68aaa259 : SUCCESS in 13m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6d280e9b73314a369427c39031fcb543 : SUCCESS in 3m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2f92e37c43c14bbea5c780f1e0324b0a : SUCCESS in 9m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/58a4e4db0a1344b5a46bec7f789a3df9 : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cb0f35e322924d4ea47628bb8b105686 : SUCCESS in 19m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3aca7fdae37e4275bd36c290c377c60b : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b8ccabf86e364aaeb921b542179435ff : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/babe130609764416924a9e19aca6963d : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f440d9218f8422aa7c2bc3772810e9b : SUCCESS in 17m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/90fe6b5f2a444b23a56cda820ca5784a : SUCCESS in 5m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1b22d8a79b464d22aa16ec90d2bf8f3b : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83821023e5c445179ea9e551951dea63 : SUCCESS in 6m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/427446f167ca49359542d458a3bd95a0 : SUCCESS in 7m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb94d1b39cce4128b71366be4b35709d : SUCCESS in 36m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1c9c31201d04ce0845e120950783ddb : SUCCESS in 1h 31m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e36afa5f478987878912183deb11b87d3bd1ae52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-28 19:45:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"82b6b87c67948f3b7e2920329b6931f75b0f75a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 19:46:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70a3e8db6c2d399c889f42ad32530dc9e87e5bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 20:52:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3906d5bc990c49e0925118cbad27a6b6 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac1273502c24d7eaa85ef990966209e : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf4972e2503f45359541dfc17e5b194c : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a36fe0df16b04995a2fe57a0722e4cd7 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59956346cf2e431dbcedbfa49dfd0ddf : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acd4d986c74d4a33a293c11afc59dbb2 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4d95d63fba24479c88b69769d7067cb4 : SUCCESS in 6m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/21f3bd0111254a878b2fb91b18683d5c : SUCCESS in 14m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4c37d1f31cdf42c0b3208ca980f5a0d3 : SUCCESS in 7m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fe467eea85b8468999cd284dd639710b : SUCCESS in 5m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cb77bd603bf34cb89c5d45b03e92be78 : SUCCESS in 5m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/83095a309c174feb97ee55d09136dba5 : SUCCESS in 8m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/80fd2c67c6a541cc95e3de1521c883b0 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/475afca5762942d0ba8229e9e2c16e5b : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1442347790c24e3c98a1345e47d41cf9 : SUCCESS in 8m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b6cafe63b6d44422abb07117ddf73039 : SUCCESS in 10m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5e393c1ac1394bf6bfda5a54c3b14b1b : SUCCESS in 3m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/db7c52acbd4c43308fe71973fe07c003 : SUCCESS in 10m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3df2ae11dfe94d8fb2ff58e3d893fba4 : SUCCESS in 5m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/55fdcfd2457d45babf67a99820f3adb6 : SUCCESS in 24m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/718ccba7e99c4e97b2fafcec6b032946 : SUCCESS in 11m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4670bb925324487b945040f82912b8c8 : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22df32d22f5743b59ed899f2796dc4ec : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25a693dfbb4e47acad48abae3b728cae : SUCCESS in 22m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/500c6ed8ab954940994f1f9d4b257847 : SUCCESS in 4m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/31f9c84102d143d4a3bfc7726b2fbc0f : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f552327e7db4de9b83e65393b4b8991 : SUCCESS in 5m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3c4b17110d654cd6bf08792152b57b9d : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eaa3e49b2a344127a1e03c496eba6795 : SUCCESS in 34m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dc72ae8ca7a42e591591eb47b9ea5fb : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":3},{"id":"59474256f223fc5ba6ebe73b43ba2d810164fc1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 20:52:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a3da4484e785eac46aaa3f48938c402bca09cc19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 20:53:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b99454eb6e4fb4b3cb0d5e4240e3fb : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fa9b729e0c96151a0ca2a82666933e78960ac7ca","revisions":{"3503a5f64cb927643d896c4a40f13a5239f50ce8":{"kind":"REWORK","_number":1,"created":"2022-05-28 13:44:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/843722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/843722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/843722/1"}}},"commit":{"parents":[{"commit":"6baaf5c4f4862e7185b731d59a9e33063de0798e","subject":"Merge \"Add Jammy wheel build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6baaf5c4f4862e7185b731d59a9e33063de0798e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-28 13:44:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-28 13:44:50.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I80d5c5bc2f379a9bac41e4bb40594a253853ee12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3503a5f64cb927643d896c4a40f13a5239f50ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3503a5f64cb927643d896c4a40f13a5239f50ce8"}]},"branch":"refs/heads/master"},"0f2e01c444ae5ef8bf28495691105a95f93840ac":{"kind":"REWORK","_number":2,"created":"2022-05-28 15:36:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/22/843722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/843722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/843722/2"}}},"commit":{"parents":[{"commit":"6baaf5c4f4862e7185b731d59a9e33063de0798e","subject":"Merge \"Add Jammy wheel build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6baaf5c4f4862e7185b731d59a9e33063de0798e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-28 13:44:50.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-05-28 15:36:26.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I80d5c5bc2f379a9bac41e4bb40594a253853ee12\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f2e01c444ae5ef8bf28495691105a95f93840ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f2e01c444ae5ef8bf28495691105a95f93840ac"}]},"branch":"refs/heads/master"},"fa9b729e0c96151a0ca2a82666933e78960ac7ca":{"kind":"REWORK","_number":3,"created":"2022-05-28 18:01:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/22/843722/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/843722/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/843722/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/843722/3"}}},"commit":{"parents":[{"commit":"6baaf5c4f4862e7185b731d59a9e33063de0798e","subject":"Merge \"Add Jammy wheel build jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6baaf5c4f4862e7185b731d59a9e33063de0798e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-28 13:44:50.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-05-28 18:01:42.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I80d5c5bc2f379a9bac41e4bb40594a253853ee12\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa9b729e0c96151a0ca2a82666933e78960ac7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa9b729e0c96151a0ca2a82666933e78960ac7ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
